)]}'
{"id":"openstack%2Foslo.log~920267","triplet_id":"openstack%2Foslo.log~stable%2F2024.1~Ia4c8e82f957c4ef3c176246fea0bc0adab4f4dc4","project":"openstack/oslo.log","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-05-23 15:23:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-05-28 11:44:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia4c8e82f957c4ef3c176246fea0bc0adab4f4dc4","subject":"Remove the usage of the Eventlet debug feature from oslo.log.","status":"ABANDONED","created":"2024-05-23 10:09:36.000000000","updated":"2024-05-28 11:44:42.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa68651ab256d3b868b299cf3af0b4f8ee9e1f72","_number":920267,"virtual_id_number":920267,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-23 11:36:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 10:25:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-05-23 11:36:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5892943d6202354a8ab6a7dedd14ca10798b1a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-05-23 10:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1562894511f32b1905a7b82336ae4d075c290036","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-05-23 10:25:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b9f4e789a6dda440ac14e14b5d9d0f285715c33","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-05-23 11:30:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2403781521a3ccf2519c36f1b52137cbe0d5f461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 11:36:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ad2d8ee209f4585a8b33017324b2a4d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11b26110d099419fa92cac2aa8afaf23 : SUCCESS in 1h 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0f8adb7dbc4110b2b159df282016c9 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a88f9701dbf46cfb73ad116e78c0cad : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ec8f86d73fe4d1baaca2e0409bf6af8 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15ff7ec178940b5a12fe24d226199f5 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded5b78e84af48b3bf91fb068a2a86ed : SUCCESS in 3m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/ec9b42d254544e689a2287940f11e4bb : SUCCESS in 1h 01m 05s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/421e038c498743b69620eba3e8cfe3af : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"71549a35cbbf9c4e06ab0b08260e6aa0435823b5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-05-23 15:23:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa68651ab256d3b868b299cf3af0b4f8ee9e1f72","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-05-28 11:44:42.000000000","message":"Abandoned\n\nMerge other patches first.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9269f6c18c8d324f5043982bdbbc6527211c282","revisions":{"a9269f6c18c8d324f5043982bdbbc6527211c282":{"kind":"REWORK","_number":1,"created":"2024-05-23 10:09:36.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/67/920267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/67/920267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/67/920267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/67/920267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/67/920267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/67/920267/1"}}},"commit":{"parents":[{"commit":"a6c4f8a9c62ae264f819716177f3da899eaf1a54","subject":"Fix eventlet detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a6c4f8a9c62ae264f819716177f3da899eaf1a54"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-05-07 10:03:59.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-05-23 10:09:07.000000000","tz":120},"subject":"Remove the usage of the Eventlet debug feature from oslo.log.","message":"Remove the usage of the Eventlet debug feature from oslo.log.\n\nWe initially started to work on this topic with the intention to solve\nthe problem of using the Eventlet hub_prevent_multiple_readers debug feature.\nThis feature is a debug convenience. This feature is not supported by the new\nasyncio hub[1] of Eventlet. You can also check the documentation here[2] from eventlet. The problem with disabling this procedure is that\nit exposes you to risk. Deactivation applies to the entire stack. If a project\nuses oslo.log, for example nova, then it exposes all threads to concurrent access\non the process file descriptors.\nAs explained in the function\u0027s documentation[3], when several greenlets are reading from\nthe same socket, it\u0027s difficult to predict which greenlet will receive which data.\nYou really need to be sure of what you\u0027re doing to use this function. Also explained\nin the raise condition added by this commit[4] and we have a reproducer[5] that exposes\nthe problem. According to our tests, removing this line does not change the fix provided.\nIt also seems that cinder no longer uses[6] logging native thread.\n\nRelated-Bug: #1983863\n\n[1] https://review.opendev.org/c/openstack/governance/+/902585/17..18\n[2] https://eventlet.readthedocs.io/en/latest/asyncio/migration.html#known-limitations-and-work-in-progress\n[3] https://github.com/eventlet/eventlet/blob/master/eventlet/debug.py#L149\n[4] https://github.com/eventlet/eventlet/commit/cb7c8c0196ed70665b0382909141ac743d7633a2\n[5] https://github.com/4383/oslo.log/blob/reproducer2/reproducer.py\n[6] https://bugs.launchpad.net/oslo.log/+bug/1983863/comments/18\n\nCo-authored-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ia4c8e82f957c4ef3c176246fea0bc0adab4f4dc4\n(cherry picked from commit 8705f67bfbd733edf923003c0a5850e7a4a5735b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a9269f6c18c8d324f5043982bdbbc6527211c282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a9269f6c18c8d324f5043982bdbbc6527211c282"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
