)]}'
{"id":"openstack%2Foslo.log~921797","triplet_id":"openstack%2Foslo.log~master~I6d1ee13409fe84cd54b7a3aa3ed862bc6e33f1c3","project":"openstack/oslo.log","branch":"master","topic":"bug/2069084","attention_set":{},"removed_from_attention_set":{"32057":{"account":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"last_update":"2024-06-26 11:44:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d1ee13409fe84cd54b7a3aa3ed862bc6e33f1c3","subject":"Replace deprecated logging.warn() calls with logging.warning()","status":"MERGED","created":"2024-06-11 19:49:13.000000000","updated":"2024-06-26 11:45:05.000000000","submitted":"2024-06-26 11:44:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921797-bug/2069084","meta_rev_id":"80b3668d9cfe04f62f940b1a867102f7ec4786ce","_number":921797,"virtual_id_number":921797,"owner":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 11:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-06-12 08:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-25 10:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-25 10:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-11 21:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 08:55:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-06-25 10:36:33.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"351c0fa35aa0fd5921fc30213f3f849a187d0376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"date":"2024-06-11 19:49:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822ce2c1e1750c4751c477d1f462539be11b27de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 21:48:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534f9c1b925d4511a53dc6673f99d1ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eacb29e1c5c74d09aa667284f94677ed : SUCCESS in 1h 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5a8e275bda42349f41dc09a8b95d90 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64cb6525f1e6451e9b603f5060a0d5d2 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26f11ad448754fc49d8ad67d58b44972 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10eb7cf0915b4b6bb383b5fe12262584 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1895639bf6b84391b6cae289fec5ceeb : SUCCESS in 2m 59s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/a7aeb8e6ff05439a9edf6f541a3aa545 : SUCCESS in 1h 02m 27s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/425118cbcc3f4bbaa1538d873242b27f : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":1},{"id":"17c090607d60b27580f49bf2faa1439282280364","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-12 08:55:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b08cda169ec4e04fd4fedff44d4d912e1661617e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-25 10:36:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ffe1a7e348f8636b100571b2e20eb9f02504b0be","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-25 10:36:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4355b6a8c0aa1ee83a57de4beb6dc1f5285b76b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c063bdbc21a2bc687230e214e3ed3d74b8595dfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 12:12:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7384373772d94348b2ccce9ae65dd173\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4eab6e4387cf411bb520d07ca4cbfb26 : SUCCESS in 1h 30m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7f3dc334974688a05ed4d9a617bf7e : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b665410c9bd40febb676e6714715429 : FAILURE in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a04ab0714939490093b1e3b1d6b40d13 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b2bb4f49de44f1b5e030b6b30bd574 : SUCCESS in 3m 14s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/61524343c8d544d58e84103121799022 : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fb068501f0c8f6013ce7a98ada2f0668b4e0fc7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-25 12:27:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68ac5b19833c7de6ff1566d3d31245efe0f020f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:12:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e29beae40c049d0973aab8f4237816f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f107c1edd91d4e009cda197cdb183853 : SUCCESS in 1h 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6204d9544d4241ceaf9a3063597ecddf : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a88bf224fae7446fb10370dc6e28242b : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/081d176656844706abd6f7974b3eb2e6 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e9dc4355a7f408491032148cce0a4eb : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8a922dd40341cea5b705b78576600d : SUCCESS in 2m 54s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/fc2651e77dc649b296d63d0086b22fd9 : SUCCESS in 1h 05m 13s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/fb19914df6ba4b5b853d1a076ee3d830 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2864e9465d7d51b9b1e9f31487a5c196162e9541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:13:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fd656e2786400848d5d267d48910ee4807e4f5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:39:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15cd4747ad1b40ce9c2435d78fd428ec\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/191b6b690e2342dc83a8716157b44852 : SUCCESS in 1h 26m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae6b993e08a449e9e91da60aca2d4a5 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67392120a6f846bab00abc0717094ef7 : FAILURE in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f8658883c3947afa7968ab3592239c8 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e7dccf91d143f6a41e66ba1ad06919 : SUCCESS in 4m 18s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/b20abc99b93e42a7a7041aaf835df701 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b34a7b3630e7c1044cd89c29aa09c0f0dfc54cc6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-06-26 08:47:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68066cd73458b1a6cffc72600c2c9b27b4896ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 10:09:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2e6576e6ba40d2bb59e0df11d6e28a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eebb318aa048418e8e5aa01030edbbbe : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8890ee5b2a406e9b5d11035d77ba25 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193fe8425173449bb1e4fd2c26694d3f : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c5d1dd87e11436eaac3d8587f1a5ca7 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b707478942464978b2ce7f4f2b5bbc7c : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e1dadc682944469bdc85d255a71184 : SUCCESS in 3m 21s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/b8d86281e5ba4b9688c589d626c0f913 : SUCCESS in 52m 56s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/17b7ff0c688748f6aa062604c8393348 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"80fe75a3c01c078ee9eab110957397365be872a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 10:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a6e776e694ce039ebe88f66c6723e30109d774b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 11:44:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9a523f03c4493eb4b97f5fe050cfa8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/071c051c6f7e474c807fe2c8b0f7c67d : SUCCESS in 1h 32m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26af59fe8a342ebb9b0051057802218 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be62128011bd4ca09db8bc4e023ceeca : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bd347dacce24758b241b6e71bcc6f50 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d465753cd6b94e738935f334aa94b7d8 : SUCCESS in 2m 52s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d2cd33c1761f4a93a5c4fabeff0c3807 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"40fa9b9cd21fe98b43341e3debc9fc884c3abc55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 11:44:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80b3668d9cfe04f62f940b1a867102f7ec4786ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 11:45:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af45a8956a140f085c43ed25e3c9cec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbaf1cac40d46808da12d3b5d7673d5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60c39a3b874f027f690b49c987daa31c37bfb0e6","revisions":{"60c39a3b874f027f690b49c987daa31c37bfb0e6":{"kind":"REWORK","_number":1,"created":"2024-06-11 19:49:13.000000000","uploader":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"ref":"refs/changes/97/921797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/97/921797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/97/921797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/97/921797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/97/921797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/97/921797/1"}}},"commit":{"parents":[{"commit":"9bee2820a260cf109379bef34386f7fd1112c4c3","subject":"Merge \"Add option to disable color\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9bee2820a260cf109379bef34386f7fd1112c4c3"}]}],"author":{"name":"Michał Górny","email":"mgorny@gentoo.org","date":"2024-06-11 19:35:58.000000000","tz":120},"committer":{"name":"Michał Górny","email":"mgorny@gentoo.org","date":"2024-06-11 19:49:02.000000000","tz":120},"subject":"Replace deprecated logging.warn() calls with logging.warning()","message":"Replace deprecated logging.warn() calls with logging.warning()\n\nReplace the deprecated logging.warn() calls with logging.warning().\nThe former were undocumented, deprecated since Python 3.3 and eventually\nremoved in Python 3.13.\n\nSee: https://docs.python.org/3.13/whatsnew/3.13.html#logging\nPartial-Bug: 2069084\nChange-Id: I6d1ee13409fe84cd54b7a3aa3ed862bc6e33f1c3\nSigned-off-by: Michał Górny \u003cmgorny@gentoo.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/60c39a3b874f027f690b49c987daa31c37bfb0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/60c39a3b874f027f690b49c987daa31c37bfb0e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
