)]}'
{"id":"openstack%2Foslo.log~921798","triplet_id":"openstack%2Foslo.log~master~I8fc022e5ad0df8ec4bd413de12106390d9dbc0f1","project":"openstack/oslo.log","branch":"master","topic":"bug/2069084","attention_set":{},"removed_from_attention_set":{"32057":{"account":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"last_update":"2024-06-12 09:17:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8fc022e5ad0df8ec4bd413de12106390d9dbc0f1","subject":"Update test_rfc5424_isotime_format_no_microseconds output for py3.13","status":"ABANDONED","created":"2024-06-11 19:49:13.000000000","updated":"2024-06-12 09:26:29.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fe208ab81faf10bca315dbae7a938d3f87902ccf","_number":921798,"virtual_id_number":921798,"owner":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-11 21:58:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-11 21:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 09:00:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"48ec36f11278d9f4a2cbc1dcfbaf6f85b1b12303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"date":"2024-06-11 19:49:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9340003a6577faadf1b8a7b66e3e8f54bfc68c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 21:58:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7946cbfa07e4a5295d2c5d694674cd1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dcaed98fcbf4264b1d3991cc4cfc39c : TIMED_OUT in 2h 04m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec2e1de892d4cd791200481b0461614 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cadc8c4535bc41f0aa59c57b0ca25682 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33783d188f9048f1a6929cbd8418876c : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6d1323fda044b0b8833f599125c4cac : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09eb83958f9e49ca8fa5687705f867f1 : SUCCESS in 3m 52s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/bf9ed5d8c1e94c6db0e7e5ab35d65d9f : SUCCESS in 53m 13s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/7426d3f39d164aa388812e38f0a3bb5b : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"38fb8163be7a411d5d94ab480d0aef9a243db4c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-12 09:00:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a82c8ca7c4a6d5e23e12cb73e3030aaf6d0f52b8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"date":"2024-06-12 09:17:29.000000000","message":"Abandoned\n\nIncorrect solution.","accounts_in_message":[],"_revision_number":1},{"id":"fe208ab81faf10bca315dbae7a938d3f87902ccf","author":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"date":"2024-06-12 09:26:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd5eb0fdf4d74150d8e3b014ef43ad0d0cbb9a5d","revisions":{"cd5eb0fdf4d74150d8e3b014ef43ad0d0cbb9a5d":{"kind":"REWORK","_number":1,"created":"2024-06-11 19:49:13.000000000","uploader":{"_account_id":32057,"name":"Michał Górny","email":"mgorny@gentoo.org","username":"mgorny"},"ref":"refs/changes/98/921798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/98/921798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/921798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/921798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/98/921798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/98/921798/1"}}},"commit":{"parents":[{"commit":"60c39a3b874f027f690b49c987daa31c37bfb0e6","subject":"Replace deprecated logging.warn() calls with logging.warning()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/60c39a3b874f027f690b49c987daa31c37bfb0e6"}]}],"author":{"name":"Michał Górny","email":"mgorny@gentoo.org","date":"2024-06-11 19:45:05.000000000","tz":120},"committer":{"name":"Michał Górny","email":"mgorny@gentoo.org","date":"2024-06-11 19:49:04.000000000","tz":120},"subject":"Update test_rfc5424_isotime_format_no_microseconds output for py3.13","message":"Update test_rfc5424_isotime_format_no_microseconds output for py3.13\n\nIn Python 3.13, the isotime format does not include microseconds anymore\nif they are zero.  Update the test to account for both possibilities.\n\nCloses-Bug: 2069084\nChange-Id: I8fc022e5ad0df8ec4bd413de12106390d9dbc0f1\nSigned-off-by: Michał Górny \u003cmgorny@gentoo.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cd5eb0fdf4d74150d8e3b014ef43ad0d0cbb9a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cd5eb0fdf4d74150d8e3b014ef43ad0d0cbb9a5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
