)]}'
{"id":"openstack%2Foslo.log~924063","triplet_id":"openstack%2Foslo.log~master~I8dfd02964fc278ac16bb311ed102a7cfc63aa059","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2024-07-31 11:32:15.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-31 11:32:15.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-07-31 11:32:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8dfd02964fc278ac16bb311ed102a7cfc63aa059","subject":"Mock time.time_ns in test_rfc5424_isotime_format_no_microseconds for py3.13","status":"MERGED","created":"2024-07-12 13:04:55.000000000","updated":"2024-07-31 11:33:46.000000000","submitted":"2024-07-31 11:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924063","meta_rev_id":"50e63c7d28de963d93578c679ad86005eb34f367","_number":924063,"virtual_id_number":924063,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 11:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-31 10:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-31 10:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-12 14:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-18 08:27:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-07-18 08:27:51.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-07-31 10:09:11.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"22f7d0eda4180c19774cc6d5503825085ee31e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-12 13:04:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93b1346d821688d3428e1ba2409fd6ccf8bbd9e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:23:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfb1dd81d9754338889e41887403ccde\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/425c4333946046cf83d17925275bb155 : SUCCESS in 1h 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d733d49f5549f6afc80d895748c38a : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/900d0a0d69b44eff91a73cde739864d1 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea66c4fb03e040fbb2f943a26588d5b8 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9091948cb53e482ea038ba05cc1e4f9e : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495badfadb5544458c00a51de7a35bfb : SUCCESS in 3m 21s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/44c33f40f9604ceabada49e822b41296 : SUCCESS in 59m 26s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a8e8cee162234ca7bb8a14a6d9d34a84 : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"642999ad8b85aa442bbbbd1f2764b8a2e7f655a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2024-07-12 15:06:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c50a65b4cf79347c7591387d5e876e0a87f19cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 16:52:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c25eb91b14c4531a5fb1825fc925f64\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b142fad67fb946b39cb24a2b9af8197a : SUCCESS in 1h 44m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2587817e344425793037e37bcaf7371 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c61d11992bf42109cccb19acf7d69ae : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac2e6b47231a48938374b77da42ffa5b : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd3f4742d11e46ab8e1974a59cbe9ffd : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26ace321c5c4a87b77feac133decbcb : SUCCESS in 2m 56s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/2f0714dc86984db592fb8e5b3a6b5ebd : SUCCESS in 46m 32s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/86ef288db6234233a028141dff4800e0 : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ecefb865e321a9f9d2049a712094ca30b15a93b1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-07-31 10:09:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eea32eb73d1e0adae545d1af1f2199e539e98df7","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-07-31 10:09:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3072b43e6fabe7a861db5022396f894a6417e826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 10:10:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48ffd94323e32d8097720f8425211512edc66df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 11:32:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805d21e508b349c9b00b1bc120374873\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e623ce2e4d8745219dc452eb9c28b440 : SUCCESS in 1h 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a1ba2b5f3147e19999e2960afe5984 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d59c8612bad41ed9e6d95fa1e7bf8fd : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1355813ba3514c4cbda17e9a8e0632be : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361afc51f7b04bb0903acbc7b3e0a979 : SUCCESS in 4m 16s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a0b48d5b89b3428e8beb84ed21869f04 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":2},{"id":"4c3a5657e38d48a8fe28b520a285ac7916b81c5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 11:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"50e63c7d28de963d93578c679ad86005eb34f367","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 11:33:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173b8964951e41eab35ac6977ef7cba9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfe3589d25a412db30b6555f6701fb8 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e762ada33fc7fea6fdc8f8acb58375b9443eb9b8","revisions":{"b480eb69d888799b72a2f009e60b15e8662d6e92":{"kind":"REWORK","_number":1,"created":"2024-07-12 13:04:55.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/63/924063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/63/924063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/1"}}},"commit":{"parents":[{"commit":"60c39a3b874f027f690b49c987daa31c37bfb0e6","subject":"Replace deprecated logging.warn() calls with logging.warning()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/60c39a3b874f027f690b49c987daa31c37bfb0e6"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-12 12:51:10.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-12 13:04:53.000000000","tz":120},"subject":"Mock time.time_ns in test_rfc5424_isotime_format_no_microseconds for py3.13","message":"Mock time.time_ns in test_rfc5424_isotime_format_no_microseconds for py3.13\n\nIn python 3.13, the logging module has moved to use time_ns() instead of\ntime() for the `created` attribute [1].\n\n[1] https://github.com/python/cpython/commit/1316692e8c7c1e1f3b6639e51804f9db5ed892ea\n\nChange-Id: I8dfd02964fc278ac16bb311ed102a7cfc63aa059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b480eb69d888799b72a2f009e60b15e8662d6e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b480eb69d888799b72a2f009e60b15e8662d6e92"}]},"branch":"refs/heads/master"},"e762ada33fc7fea6fdc8f8acb58375b9443eb9b8":{"kind":"REWORK","_number":2,"created":"2024-07-12 15:06:41.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/63/924063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/63/924063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/63/924063/2"}}},"commit":{"parents":[{"commit":"60c39a3b874f027f690b49c987daa31c37bfb0e6","subject":"Replace deprecated logging.warn() calls with logging.warning()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/60c39a3b874f027f690b49c987daa31c37bfb0e6"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-12 12:51:10.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2024-07-12 15:06:31.000000000","tz":120},"subject":"Mock time.time_ns in test_rfc5424_isotime_format_no_microseconds for py3.13","message":"Mock time.time_ns in test_rfc5424_isotime_format_no_microseconds for py3.13\n\nIn python 3.13, the logging module has moved to use time_ns() instead of\ntime() for the `created` attribute [1].\n\n[1] https://github.com/python/cpython/commit/1316692e8c7c1e1f3b6639e51804f9db5ed892ea\n\nChange-Id: I8dfd02964fc278ac16bb311ed102a7cfc63aa059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e762ada33fc7fea6fdc8f8acb58375b9443eb9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e762ada33fc7fea6fdc8f8acb58375b9443eb9b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
