)]}'
{"id":"openstack%2Foslo.log~932290","triplet_id":"openstack%2Foslo.log~master~Ide9872a6286034035b02fa2eb472cff8837cbef3","project":"openstack/oslo.log","branch":"master","topic":"fix-tox-minversion","hashtags":[],"change_id":"Ide9872a6286034035b02fa2eb472cff8837cbef3","subject":"Fix outdated tox minversion","status":"MERGED","created":"2024-10-13 12:36:25.000000000","updated":"2024-10-13 16:12:51.000000000","submitted":"2024-10-13 16:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932290-fix-tox-minversion","meta_rev_id":"391e3c23220766d14e5a0e150f36c81cfe83e418","_number":932290,"virtual_id_number":932290,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-13 16:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-13 12:43:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-13 12:43:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-13 14:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9d4d5f424fb4f05cba451a53b9d9c7ab4bd996b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"862938c0bfd6f738db4f9a4084b3d882b31e115e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:43:17.000000000","message":"Topic set to fix-tox-minversion","accounts_in_message":[],"_revision_number":1},{"id":"5b7fbd927160a00a3a4ba34c6d4f34ecf6e136c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:43:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31ef1217a9f2f096f2aa0fc77339f46f1145bdef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 14:16:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80253d79d5504ead8150a4a731fe7237\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/736bf90020374e86a6fa087c4ab279a4 : SUCCESS in 1h 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf64432c3ea403a927b2c0db65d22df : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d3b90d78f0c4893bd9d7f4fdb14729a : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4774967211a441ce95c4028535ff6af3 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f72ef892d0247419f26e71ea2e044f5 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33fa059e7a134c2b92c1768f42c63888 : SUCCESS in 2m 26s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/9726952c30f944d59c08c7ef546d24ea : SUCCESS in 1h 05m 51s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/beee3c7fac25473e98eb8b41dea77a72 : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"268caaff4b7dcc1c2fc7a339a265102d77bdc493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 14:16:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5afd0dbd635fd173e5e9481c84074c801f6003e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 16:11:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bfc6658d17a4d1e8a1fb8c683394319\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/379a45bce3ac46029959818a67f9e6cc : SUCCESS in 1h 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dee688c2dc04dfeb802d915c84cbbd8 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d2163edd1543b1ba7e44e0b717942d : SUCCESS in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/594617725985433eb3f557dedf54ac84 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fcfd366eb994a149b34ac6d5460dc1d : SUCCESS in 2m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae1e014a438b478c9db298747669834d : SUCCESS in 3m 24s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3670da0d31ec44e0bc7626819eceb012 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b557b08095cd7daccebb1d646258b09f81e9afbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 16:11:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"391e3c23220766d14e5a0e150f36c81cfe83e418","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 16:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d8ea2bb66449959e1b296350a9e511\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b564745bc409416fb69c566785be5f85 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8dd5afabec74b13bfda4beb34d5aefa : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","revisions":{"0f5179e170c9832fa1009fd09ac4b3498b7d2404":{"kind":"REWORK","_number":1,"created":"2024-10-13 12:36:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/932290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/90/932290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/932290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/932290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/90/932290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/90/932290/1"}}},"commit":{"parents":[{"commit":"a38a01cefafe16a51452e3deb1cf533aeb349370","subject":"Declare Python 3.12 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a38a01cefafe16a51452e3deb1cf533aeb349370"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-13 12:36:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-13 12:36:16.000000000","tz":540},"subject":"Fix outdated tox minversion","message":"Fix outdated tox minversion\n\nThe allowlist_externals option is supported since 3.18.0 .\n\nChange-Id: Ide9872a6286034035b02fa2eb472cff8837cbef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
