)]}'
{"id":"openstack%2Foslo.log~932465","triplet_id":"openstack%2Foslo.log~master~Ifdd592009368e925c3636794b1548af13d7fed76","project":"openstack/oslo.log","branch":"master","topic":"remove-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-23 14:25:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifdd592009368e925c3636794b1548af13d7fed76","subject":"Remove Python 3.8 support","status":"MERGED","created":"2024-10-16 01:26:54.000000000","updated":"2024-10-23 17:10:41.000000000","submitted":"2024-10-23 17:09:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932465-remove-py38","meta_rev_id":"5e3bb6223cfbb46636f175373de04a4387046075","_number":932465,"virtual_id_number":932465,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 17:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-23 14:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-23 14:25:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-16 03:13:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33c5c16df119fb1b2d156afcda4690380b88c134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 01:26:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d19c82a93335f82b518be8a879a463b38c3d4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 03:13:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9719e2c6e19453fad45fa0934cb6223\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/921190a4678f40dea4d2d9e69655dbd9 : SUCCESS in 1m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebf7a72d3bb7400f9fc2e6d6e0a5c534 : SUCCESS in 1h 46m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74eda7cef697462e89dee1e0bb1a1eb3 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f02f27c19b9a44059aa38307c5444c06 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78129f478b9f4045afefd5e5ceb7f6ee : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daeed84d445e47ae99afdb4fed79b7eb : SUCCESS in 2m 39s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5b939b4de90247ea9cdfc78fbb72a327 : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0674e8a7ad1a3242215f2b01ce0e99940dd33ef2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 14:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87fe9a4ac3e8bec150e87e921f187dc8038e3d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:18:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd3a94740c6139a4fe1c78c5a3cd35f07a7db2aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 14:20:13.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a19ea4dc2ff988de6ea5677d78cf696946271f27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:20:26.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"dc05a957b8e41b6a035c187fc8cf8b96a5e58b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 14:21:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-2\n","accounts_in_message":[],"_revision_number":2},{"id":"3b42b56f768fda1b64f7be3353f8c9b059e2da34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 14:25:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3896e4fcfe52c23d55dd24bf69d60675f504a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 15:37:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607a6d9788074e0c99fce6a8fc0a3e33\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e86ef075a85043e9a09f3bb92330eb70 : SUCCESS in 1m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49845889651e49499b0c621e2ae65467 : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9c793f08084ecaa3e55825d0ddc13c : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a934129e29d040058a539f8d1980e02a : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11641ddb22d8454097160f7de88f5843 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b02551821294c7daa94143abb5dd2aa : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99cdbd6465924390ad51589320bbf68d : SUCCESS in 2m 57s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/47f38e7191ef4bf8abbb3b9ed6dddd67 : SUCCESS in 49m 35s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/221891240dbb49808f5245cc4c59e33a : SUCCESS in 27m 55s","accounts_in_message":[],"_revision_number":2},{"id":"31233220566e03de17789665a994c98424258540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 15:37:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ffe75c806e3c4eba52bda5416911178e8ee61a1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 17:09:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/241703a33ec7400692dc03a60264e6ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acd397b3f1384e3fa60ef20f2fdfe47c : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec6d42da7a694e83817c650a6b222438 : SUCCESS in 1h 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b489c175d540a4972e60951f4f8400 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4fe2a318254effa366d0cf38a16273 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8b78e8efcda427987dee43d21a9affa : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2575f0d683fb4e16ba9ca4a046a4a935 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a20a7e58848406789f1d30fc90c1d66 : SUCCESS in 2m 20s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/8b1a98f55e3f4213b17173736e2585a6 : SUCCESS in 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"449a6bf0b96116810cde11242cb9229346b5ffee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 17:09:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5e3bb6223cfbb46636f175373de04a4387046075","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 17:10:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ed0ab11a32d419baf40c7b8713d6317\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49453429ae8b4c36a66b4825e83c6c84 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5371f98a060f48549667309c5783a16d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dc7dde7e1d3cfe38bdd6f9b2c1e43b9163196d5","revisions":{"1f112b5bea4082713859dbb0cd3e6a25e7a68508":{"kind":"REWORK","_number":1,"created":"2024-10-16 01:26:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/932465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/65/932465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/1"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:26:44.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: Ifdd592009368e925c3636794b1548af13d7fed76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1f112b5bea4082713859dbb0cd3e6a25e7a68508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1f112b5bea4082713859dbb0cd3e6a25e7a68508"}]},"branch":"refs/heads/master"},"1dc7dde7e1d3cfe38bdd6f9b2c1e43b9163196d5":{"kind":"REWORK","_number":2,"created":"2024-10-23 14:21:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/932465/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/65/932465/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/65/932465/2"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:26:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-23 14:21:44.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: Ifdd592009368e925c3636794b1548af13d7fed76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1dc7dde7e1d3cfe38bdd6f9b2c1e43b9163196d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1dc7dde7e1d3cfe38bdd6f9b2c1e43b9163196d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
