)]}'
{"id":"openstack%2Foslo.log~933186","triplet_id":"openstack%2Foslo.log~master~I131d963bab07eff4900f6ac79ca0386f0b0159bf","project":"openstack/oslo.log","branch":"master","topic":"pre-commit-skip-functional","hashtags":[],"change_id":"I131d963bab07eff4900f6ac79ca0386f0b0159bf","subject":"Skip functional tests on pre-commit config  update","status":"MERGED","created":"2024-10-24 01:16:17.000000000","updated":"2024-11-11 17:30:23.000000000","submitted":"2024-11-11 17:28:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933186-pre-commit-skip-functional","meta_rev_id":"1e3bc430ae03af706d8478c618e40fbeb70a7b1d","_number":933186,"virtual_id_number":933186,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-11 17:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-11 15:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-11 15:54:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 04:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11a1c9b7d60e12b88800150ba6953b43b15bcfd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:16:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"894345e16f84fd28006bf50f081cb7621c564acf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:17:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"22532172044dbf9d9ea02ed1e4324afc2e536bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:24:56.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"60a3e4de2870ec17c626aa4ff55eb19d89ff6ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:40:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b24c686584750e89a0295d31de3bc9948f3f5c93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:41:33.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"048c0376aea056f806224875b769a9f4f01843d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:41:42.000000000","message":"Topic set to pre-commit-skip-functional","accounts_in_message":[],"_revision_number":5},{"id":"e1f79802b1da883b45e681eb729c77917f1c0288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 03:25:41.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"103201848faac9b1e3434c1bfdfddbab1c1e997b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 04:40:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fffc94a59b4d1d96653219e3f807e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d180124dfd2a44d3906c6df09847a07b : SUCCESS in 1h 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769d58b3f6d94e61afbb60a3119a5b61 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95921fe465b14803b98249b64b1df0a3 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0df6f3dd82884745afa0c623ddb629fa : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833d1671b028402ba06f18876f86ac96 : SUCCESS in 2m 39s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/782913b0f05046d5888d59ae9b3f7c63 : SUCCESS in 43m 26s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a90ff608f4b044f5844dee1659b3c946 : SUCCESS in 22m 56s","accounts_in_message":[],"_revision_number":6},{"id":"e8c7e5aa93da0566a2371e278578ee8243646b2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-11 15:54:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"14202ee4d73ec239b95b91e72df8fd706fbd3636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 15:54:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9901da008821ce01c3671f11ba768da08bc906bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 17:28:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec02fe574a3542a590e994a900d0c34d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b8f2108bc2a45a4b5553e5f937cb60d : SUCCESS in 1h 28m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50d5d79dd0a4cd1a4fba9f8f80ae59f : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0f31192a274870ae0bf8981c24aff0 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0de623d6964421cbea079a1307a6500 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a5047795e84cb197507747542dcc65 : SUCCESS in 2m 51s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d56efd867b2840dd877eab08d27781fd : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":6},{"id":"baa1fc609ee497b4330cb61dc6a9c07215c9876c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 17:28:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1e3bc430ae03af706d8478c618e40fbeb70a7b1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 17:30:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e0c2d3dec743b99721af4f1d812ae7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e335a41350374b18a1301bae5bbbf078 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f480f63104d09933783af90e914278340c92c2d7","revisions":{"6f6b7fe0bed455b1f667319d70f9a2ae3e72a84a":{"kind":"REWORK","_number":1,"created":"2024-10-24 01:16:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/1"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:16:05.000000000","tz":540},"subject":"SKip functional tests on pre-commit-config.yaml update","message":"SKip functional tests on pre-commit-config.yaml update\n\nThis file is used only in pep8 .\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f6b7fe0bed455b1f667319d70f9a2ae3e72a84a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6f6b7fe0bed455b1f667319d70f9a2ae3e72a84a"}]},"branch":"refs/heads/master"},"30fbc5cd186df0884a76c74fd4bce308c488b71d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-24 01:17:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/2"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:17:08.000000000","tz":0},"subject":"Skip functional tests on pre-commit-config.yaml update","message":"Skip functional tests on pre-commit-config.yaml update\n\nThis file is used only in pep8 .\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/30fbc5cd186df0884a76c74fd4bce308c488b71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/30fbc5cd186df0884a76c74fd4bce308c488b71d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"783cb8d9f0060844f56cd717468ace72bc59ed0d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-10-24 01:24:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/3"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:24:56.000000000","tz":0},"subject":"Skip functional tests on pre-commit config update","message":"Skip functional tests on pre-commit config update\n\nThis file is used only in pep8 .\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/783cb8d9f0060844f56cd717468ace72bc59ed0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/783cb8d9f0060844f56cd717468ace72bc59ed0d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eb15ad711ccf681e01802840e4c81adad0e60f0f":{"kind":"REWORK","_number":4,"created":"2024-10-24 01:40:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/4"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:40:16.000000000","tz":540},"subject":"SKip functional tests on pre-commit config  update","message":"SKip functional tests on pre-commit config  update\n\nThis file is used only in pep8 .\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eb15ad711ccf681e01802840e4c81adad0e60f0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/eb15ad711ccf681e01802840e4c81adad0e60f0f"}]},"branch":"refs/heads/master"},"6d339ddedb1aa3ec17f90b5c2fd2ad03ede3221d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-10-24 01:41:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/5"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:41:33.000000000","tz":0},"subject":"SKip functional tests on pre-commit config  update","message":"SKip functional tests on pre-commit config  update\n\nThis file is used only in code style checks\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6d339ddedb1aa3ec17f90b5c2fd2ad03ede3221d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/6d339ddedb1aa3ec17f90b5c2fd2ad03ede3221d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f480f63104d09933783af90e914278340c92c2d7":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-10-24 03:25:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/933186/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/86/933186/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/86/933186/6"}}},"commit":{"parents":[{"commit":"0f5179e170c9832fa1009fd09ac4b3498b7d2404","subject":"Fix outdated tox minversion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0f5179e170c9832fa1009fd09ac4b3498b7d2404"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:15:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 03:25:41.000000000","tz":0},"subject":"Skip functional tests on pre-commit config  update","message":"Skip functional tests on pre-commit config  update\n\nThis file is used only in code style checks\n\nChange-Id: I131d963bab07eff4900f6ac79ca0386f0b0159bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f480f63104d09933783af90e914278340c92c2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f480f63104d09933783af90e914278340c92c2d7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
