)]}'
{"id":"openstack%2Foslo.log~935800","triplet_id":"openstack%2Foslo.log~master~Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b","project":"openstack/oslo.log","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-11-23 13:46:20.000000000","reason":"Change was submitted"},"36603":{"account":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"last_update":"2024-11-23 13:46:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b","subject":"Support running tests w/eventlet monkey patching","status":"MERGED","created":"2024-11-20 15:06:04.000000000","updated":"2024-11-23 13:48:33.000000000","submitted":"2024-11-23 13:46:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"935800-eventlet-removal","meta_rev_id":"278fb0716a7e89f698f347293fca9cc4d614aec7","_number":935800,"virtual_id_number":935800,"owner":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-23 13:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-11-23 12:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-22 09:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-11-23 12:10:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 15:26:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-11-20 16:00:20.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-11-20 16:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 09:12:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-11-23 12:10:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1469811696c738f4301f9909991f872dbd01fe82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"date":"2024-11-20 15:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0bd6791cbc55918af7a89347199a51c20b176a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"date":"2024-11-20 15:06:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6b37c7dcb334d03ce5e5f8288347faf86e88db9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-20 15:26:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6addd7907c9dbee73e35b37b1c2b998c3cf7ac8","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-20 16:00:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"41f25df753f9a0342cc9f4f1cbbcbbcf8598af66","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-20 16:03:45.000000000","message":"Topic better-testing-and-deadlock-fix removed","accounts_in_message":[],"_revision_number":2},{"id":"8d45d1300beb93ad89c9993107a70c86be0382f0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-20 16:03:56.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":2},{"id":"1e281838696c31d18d2632cd6a25efe4a504df31","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-20 16:09:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3810f23ba76bd6e81efbd2897065f897fdb1d11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:48:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4ddc91d42d479abd5ee30d6b892369\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71985bc794f3438d92e95ab68203e648 : SUCCESS in 1h 39m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eea60fbbfe4f3892434f0ec0b94a49 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e73a1494be7d49ac91245d714e718c77 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4703de05a6a4809abc1c8a331b0d8df : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd7b75bc02b47c6aea727b91c52a952 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64381590395d4d529d36a9acb1017156 : SUCCESS in 3m 34s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/04b5ac0d959948829958f75b6da42a77 : SUCCESS in 55m 29s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/cd629af71d7346c1aafb89b2d1d9716c : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":2},{"id":"927303b384b43118c6aa9ed63cdacea7389fbaba","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-20 23:55:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3320ad94ee027028462914238d9f99945175ace","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-21 09:49:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2d8943ab794b14107b73d8ed8ef45b567b63355","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-21 09:56:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bd7666bd24aba00c71967f1836f1b5e2dc8bdca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-21 10:02:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6c945ca5359dc63440b3108111f3c74d49df335","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-21 10:04:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b9f3e3ffff98716d402889534be95d083cf18d5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-21 17:47:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5c5d0c899491cf9a5b05e0ca35979a389bce9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-21 21:48:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dfd41152972e6788541c9c2138493d4eee84ab0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 21:48:23.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/oslo.log on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job: null\n  name: oslo.log-tox-py312-eventlet-patched\n  parent: openstack-tox-py312\n  vars:\n  ...","accounts_in_message":[],"_revision_number":3},{"id":"2b1ae003adb4b554f8908d545b26787b52ef7e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-21 21:51:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"543d391b4e6006607ba12a104a6582734f79ad35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-21 21:59:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6dba2e9386e107b8a6554fdd4f9b8074c1235efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-21 22:37:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"159ce95df9871edfe20f23e409e34a0ff3739450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 00:14:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d381b96bd5a14446991bbc1d5e859567\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73a03f497e1349cbbd6eafe8024e7e23 : SUCCESS in 1h 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9059b30391349ecbc61cb288d69dbf7 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eaede26412a415ea20e4469b5c4b426 : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e527b5ac5f95499e8c37e1a285fada03 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016bb1ea6ff14afebca44e063123f01c : SUCCESS in 2m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ad1dfac7da64caeb74ebc27a897fc2d : SUCCESS in 3m 45s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/e7cd84a746454e4889389d29cf996916 : SUCCESS in 1h 10m 14s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6bcb9ea569d641ebaeeb867e011c306b : SUCCESS in 25m 52s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/1643e6936c87486f8f4eae89c98922c0 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"72693a30f4f79bd6d607f510e5fa519cc0a9a1df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-22 09:12:31.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea4d6b59552d0e660c816f5009193d148e657cd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 12:10:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d2fc50258498ea67bd8384060c01e04c3c48d12a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 12:11:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d634bd47626fe0b74164ddecb8ab06786c805028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 13:46:20.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848e1260c9a34c73a3afb5841b88ae7c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b5aaa68a650447cb48bb1a3bc553dae : SUCCESS in 1h 35m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48fb467c06e746c48ed2614e24417368 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92f15f9f7a8f4af087893713e8610251 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1411d962d5a443a3952f73db0e7eb1be : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae00ac77408425496046db60246074e : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b7b32bdc394956aaf417687e1347f3 : SUCCESS in 2m 03s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/18f08c6bf996490d85b37cda82228e92 : SUCCESS in 20m 30s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/4abbb41d967d4e9787e66cd84c5d9610 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":6},{"id":"818026e2bba6d9c8792247cd8c2cba1d3f8483f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 13:46:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"278fb0716a7e89f698f347293fca9cc4d614aec7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 13:48:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae5854168c04e71a4be4190062dae48\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01190e4bab2a46bdaadd72b565b14996 : SUCCESS in 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae33ba3a7f704e8a9f00477705389556 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"98bb185f15d1d74a40d7473887d7c57b966ae048","revisions":{"885f02c80e32a702c70cb9a722c5426829e957fa":{"kind":"REWORK","_number":1,"created":"2024-11-20 15:06:04.000000000","uploader":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"ref":"refs/changes/00/935800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/1"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-20 15:06:00.000000000","tz":-300},"subject":"Support running tests with eventlet monkey patching, and fix deadlock this found.","message":"Support running tests with eventlet monkey patching, and fix deadlock this found.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/885f02c80e32a702c70cb9a722c5426829e957fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/885f02c80e32a702c70cb9a722c5426829e957fa"}]},"branch":"refs/heads/master"},"178d93e978d0f513ba08a32fcf8c29fdf9d613ab":{"kind":"REWORK","_number":2,"created":"2024-11-20 15:06:59.000000000","uploader":{"_account_id":36603,"name":"Itamar Turner-Trauring","display_name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","username":"itamarst2"},"ref":"refs/changes/00/935800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/2"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-20 15:06:52.000000000","tz":-300},"subject":"Support running tests with eventlet monkey patching, and fix deadlock this found.","message":"Support running tests with eventlet monkey patching, and fix deadlock this found.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/178d93e978d0f513ba08a32fcf8c29fdf9d613ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/178d93e978d0f513ba08a32fcf8c29fdf9d613ab"}]},"branch":"refs/heads/master"},"82cdea93064c0f6d932d01c7a2b0ece7ca45a539":{"kind":"REWORK","_number":3,"created":"2024-11-21 21:48:01.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/00/935800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/3"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-11-21 21:48:00.000000000","tz":-480},"subject":"Support running tests w/eventlet monkey patching","message":"Support running tests w/eventlet monkey patching\n\n- Enables the ability to monkey-patch eventlet during testing\n  by setting OSLO_LOG_TEST_EVENTLET\u003d1 in environment.\n- Fix deadlock running tests under eventlet monkey-patching\n- Enable new CI job which runs unit tests with evenlet monkey\n  patching enabled.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/82cdea93064c0f6d932d01c7a2b0ece7ca45a539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/82cdea93064c0f6d932d01c7a2b0ece7ca45a539"}]},"branch":"refs/heads/master"},"4340b1ff508db0bce24b1383ac3c98ecd76d4f91":{"kind":"REWORK","_number":4,"created":"2024-11-21 21:51:04.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/00/935800/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/4"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-11-21 21:50:57.000000000","tz":-480},"subject":"Support running tests w/eventlet monkey patching","message":"Support running tests w/eventlet monkey patching\n\n- Enables the ability to monkey-patch eventlet during testing\n  by setting OSLO_LOG_TEST_EVENTLET\u003d1 in environment.\n- Fix deadlock running tests under eventlet monkey-patching\n- Enable new CI job which runs unit tests with evenlet monkey\n  patching enabled.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4340b1ff508db0bce24b1383ac3c98ecd76d4f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4340b1ff508db0bce24b1383ac3c98ecd76d4f91"}]},"branch":"refs/heads/master"},"70cb61555b67108df7e90a359ab736bc3b0a840f":{"kind":"REWORK","_number":5,"created":"2024-11-21 21:59:19.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/00/935800/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/5"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-11-21 21:59:00.000000000","tz":-480},"subject":"Support running tests w/eventlet monkey patching","message":"Support running tests w/eventlet monkey patching\n\n- Enables the ability to monkey-patch eventlet during testing\n  by setting OSLO_LOG_TEST_EVENTLET\u003d1 in environment.\n- Fix deadlock running tests under eventlet monkey-patching\n- Enable new CI job which runs unit tests with evenlet monkey\n  patching enabled.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/70cb61555b67108df7e90a359ab736bc3b0a840f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/70cb61555b67108df7e90a359ab736bc3b0a840f"}]},"branch":"refs/heads/master"},"98bb185f15d1d74a40d7473887d7c57b966ae048":{"kind":"REWORK","_number":6,"created":"2024-11-21 22:37:00.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/00/935800/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/00/935800/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/00/935800/6"}}},"commit":{"parents":[{"commit":"76b50be7b1e280af2ec2e67574cf2496af0d54b7","subject":"Merge \"Skip functional tests on pre-commit config  update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/76b50be7b1e280af2ec2e67574cf2496af0d54b7"}]}],"author":{"name":"Itamar Turner-Trauring","email":"itamar@pythonspeed.com","date":"2024-11-18 17:38:33.000000000","tz":-300},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-11-21 22:36:52.000000000","tz":-480},"subject":"Support running tests w/eventlet monkey patching","message":"Support running tests w/eventlet monkey patching\n\n- Enables the ability to monkey-patch eventlet during testing\n  by setting OSLO_LOG_TEST_EVENTLET\u003d1 in environment.\n- Fix deadlock running tests under eventlet monkey-patching\n- Enable new CI job which runs unit tests with evenlet monkey\n  patching enabled.\n\nChange-Id: Ie9e012ac73bbfa8b5e6e0f610c11868466c7cf4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/98bb185f15d1d74a40d7473887d7c57b966ae048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/98bb185f15d1d74a40d7473887d7c57b966ae048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
