)]}'
{"id":"openstack%2Foslo.log~936083","triplet_id":"openstack%2Foslo.log~master~Iad88ba09bd720b7df94f1f0cd9654238bb5668ba","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-02 03:07:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iad88ba09bd720b7df94f1f0cd9654238bb5668ba","subject":"Skip grenade job for doc update","status":"MERGED","created":"2024-11-23 12:34:18.000000000","updated":"2024-12-02 07:05:34.000000000","submitted":"2024-12-02 07:02:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936083","meta_rev_id":"fe112371f18c162033d9179c6595a75c680ca0e7","_number":936083,"virtual_id_number":936083,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-02 07:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-11-25 21:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-25 15:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-01 23:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-23 14:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-25 15:02:35.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-11-25 21:57:36.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"94fd1a7f9574ad584c63c4828c54a398adce3c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 12:34:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d7b50ad7455d7381d53e6cefcb7b5238bee5d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 14:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d35827f05b4437093a16ef86764f825\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4896b50df63c4e1e9468452ed682cdef : SUCCESS in 1h 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cca5a08ae3b4a508d314d4ff8422c77 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a693f7984b184335989e40ffe20608c4 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04d5187838984a5bb692d2125fb36f8a : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9132dcd0354f78b8ca0d5e66d482fb : SUCCESS in 2m 26s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/43bca68f08244c14992093dae5464ce1 : SUCCESS in 52m 03s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2eb92f6daa2f442d85c72b2764e58212 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b299082724dfa6a2003f1bf151210810ee12bb99","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-25 15:02:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49aa76cc809e8b301b40ce777cb99fe1346e05f2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-25 21:57:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2f2b3159db347d7c3855046d83396375fddcaf5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-12-01 23:02:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d5501267798efb4fb318529fa9670305f887d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 23:03:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de8fb25d1fb798298742e982599acdea84b61fdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 00:03:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5e243aa3d314fb0b035440b9b8bcea7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/131c3dfddae044ca92a5aa73d455e81e : FAILURE in 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244c028f9eae4c2c8dc997d643deecd2 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e38c369e7784822bb4026a5b4792e54 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b055e74cf0494e7cb3ff04acfe7dbe50 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97334ecab1154d9e80d45da915948b6f : SUCCESS in 2m 50s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/85d152fd367645c0806fe294afba9af5 : SUCCESS in 22m 10s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/0e9bb0ba6e4c445194a6441be81b29d8 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fbc763a4c897375caae3604e28db0270d9627209","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-02 03:07:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39ca46f7156f87b6f08f15adb1dbe01d693c5a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 05:16:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fab27c8e34c41bab9e95ab4e6084dab\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2d345450d2a43cd959248d9e93d5849 : SUCCESS in 2h 02m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8aa39afce194354a55dc6b169c2faa5 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cddda8636f094358abf2695f9c5daa29 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27b6e5ab26bd456b962a558b32ba8e81 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a75fbd6e0b4805a7feb6b2a48890a8 : SUCCESS in 2m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/99818b449d4d4d57b9bace6a038a0f9e : SUCCESS in 49m 22s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/762be580b64d43adbb546121bc75483c : SUCCESS in 28m 26s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/ba1cae573cd94aa3a43528b6a0b5a1e0 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a251e61cf982091d37b4d6a18b7d9a80a48907d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 05:16:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ee25b9536bc4d5e439aae8d1993b2251e4b0d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 07:02:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1683f48067a6448bbe12fe2ebb9469c8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e39966b0077454f97ae4f194396bc53 : SUCCESS in 1h 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830151764e8b43a69b98079744298027 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1738f9e56a0a4c21a448a14f6695fa00 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b8d221e878d47648d41ce6543454ce4 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d66fb7808d4109a20a8aaaad0acce4 : SUCCESS in 3m 31s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/3700a7a9a1ad4d1db5a05910b3a88e65 : SUCCESS in 41m 44s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/333379e3253643f2960fc14124c1eb54 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"54e8bc41b13cdb0d5c9f9342d800ed6ffce2a611","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 07:02:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe112371f18c162033d9179c6595a75c680ca0e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 07:05:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd7b002b401f4a289f0010a336931beb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e0812a35814185b3f9e2982255f812 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1290441e3039512cb1ee24953e16806dec9eef6d","revisions":{"1290441e3039512cb1ee24953e16806dec9eef6d":{"kind":"REWORK","_number":1,"created":"2024-11-23 12:34:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/936083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/83/936083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/83/936083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/83/936083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/83/936083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/83/936083/1"}}},"commit":{"parents":[{"commit":"68903297d7534aae4b65173788d74c8273610835","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/68903297d7534aae4b65173788d74c8273610835"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 12:33:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 12:33:48.000000000","tz":540},"subject":"Skip grenade job for doc update","message":"Skip grenade job for doc update\n\n... and also remove the reference to api-ref directory which does not\nexist in this repository.\n\nChange-Id: Iad88ba09bd720b7df94f1f0cd9654238bb5668ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1290441e3039512cb1ee24953e16806dec9eef6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/1290441e3039512cb1ee24953e16806dec9eef6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
