)]}'
{"id":"openstack%2Foslo.log~940432","triplet_id":"openstack%2Foslo.log~master~I35797e0d03b69d77733a49c80b6e06c964fb48bf","project":"openstack/oslo.log","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-06 18:22:51.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-02-06 15:11:03.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-02-06 18:22:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35797e0d03b69d77733a49c80b6e06c964fb48bf","subject":"deprecate eventlet monkey patch within oslo.log","status":"MERGED","created":"2025-01-30 09:52:54.000000000","updated":"2025-02-06 18:23:54.000000000","submitted":"2025-02-06 18:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940432-eventlet-removal","meta_rev_id":"5f58b1ddcba0284fe5761fc6fce945db0cf2132c","_number":940432,"virtual_id_number":940432,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-06 18:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-02-06 15:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-02-06 15:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-02-06 15:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-30 11:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-30 16:41:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-02-02 14:57:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-02-06 15:10:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d43682210dfc35a66c9a21d963b692706bfc92db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-01-30 09:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9682bc9fde84f977eaa28d2c3c2ba71701cd013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 11:48:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f6b518975b42f8aa46a82b8c0801d1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/63c9ce4ec4134d0a900ae8cd907824bf : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bcb4794051a4250b78e3b46b97ab8e9 : SUCCESS in 1h 48m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6aa91abad6f40f7bdd1346801325271 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e157cc34d40449db18efe27f812b8be : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5d36f2b66cb4cf196e3a0c179a7ef62 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41668e72a8e84bf9a853243ce927e939 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d95b67ea96847309cb6b6839e14012b : SUCCESS in 2m 53s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/0737eece970a459e8f16d010c3825022 : SUCCESS in 38m 50s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ca96392b6a28485b8c06e60847b7a204 : SUCCESS in 43m 34s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/256988f0bc9d4fe688809a3a1e93473a : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6ebd86445b6efaf15879186903554dd213e7e2d8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-01-30 16:41:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6879752daf31e5aab75f5ac535cf2c954de7d638","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-02 14:57:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"050ae50a61ef32b468f3e2fbda50dea06c8750aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-03 09:35:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"038ad45a61aae159e0e268ef6ae8d1246b9c548c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 11:27:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38a6139702924fdc84609211f923bf0e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7765e670edb4c94bd606479ad61a366 : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e69c57de554336b9b698f3646776c5 : SUCCESS in 1h 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ededcc6fff445d9eb18103a6984098 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/088f84aeb8aa4e57a5a12c43170eb716 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c1646c0c4af433b9b023c065940c2df : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac9469735da4a18aaf62394280e69b0 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f80ee3d5c68c44cb8b3e23fc51f947cb : SUCCESS in 3m 27s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/b39bf64025594693afa03d1898a02513 : SUCCESS in 1h 04m 55s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6f46face12964de487850cd8c2d24dc2 : SUCCESS in 22m 59s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/57fc463843e444c1895721880bf4c3f0 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"43574337abd5b8c4c61cd2527c682228788884ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-05 12:32:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9210d049d37c9562594ac4a5f82b0a0392049686","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-02-05 17:45:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56d5d7a4610142570e999c9a20dc23313f50e81c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-06 15:10:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"999e7a5226c95fd0235bc784902822ec9780390b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-06 15:10:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cd9af6bdaf864195d13971db8023429dc7d701f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-06 15:11:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d1bd81befc320309ebaae1fc142e2affd8f3942a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:44:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2fda5154b14bcab240ea1de3e78a37\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c1efa762f60e482a89462aeef33badf7 : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6499905fb2a482b9a0413a0fb2c7db0 : SUCCESS in 1h 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c15d0c2089641d9b376b616d8896ece : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60a070c9ca44a6e806ef5f4d69f4bad : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b7809b2f7144b55a44d7f16226518e1 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716c91f93f214b428e4dbeacb89a87d5 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85e9984688954504bee190b4613b4fef : SUCCESS in 2m 26s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/3e91e73714384d48b8c249e8757bf553 : SUCCESS in 1h 00m 01s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/88b1f30e588643528c0f51662228b370 : SUCCESS in 24m 59s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/88c13f76d7d640bdaeebffc9a95e116d : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d9b8721799fb77603f18bc4b8f4d3872444fee4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:44:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3bf65d1a626084a4d611da0b61c51d1087adea56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:22:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2524491bd74e82bfc640e4ddaf184f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b761dc76e9a24c10b731ef638ba2b39f : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c0fc78256994ff4844f8c7acce71336 : SUCCESS in 1h 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a854350eed4ac797a5ec27cff2ec02 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f249e581b83540d1a6b2322b07bd23c0 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a16f4207813460187eb3023d16886df : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8f5dbb069e46f7b3fa711bd19809de : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c703287b0e54dea9c363eda5ab6ebfe : SUCCESS in 3m 21s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/dde628f3f7b5413cb3cb4f86e70db12b : SUCCESS in 44m 47s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/cc84c95913d544a8936cd2c430c6612b : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1b8630a879b68e0a3c478c2b9ab52aad45360f85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:22:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5f58b1ddcba0284fe5761fc6fce945db0cf2132c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:23:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21ddf88858e4ce4b129eb4d96e834f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5962f6f3fb447a99559025185b2cca : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76bdae1661e0440e8ad5b5477af23d0e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e9730dacb2a3915d926863873895b1f28280ca5e","revisions":{"bf240676f1a8cdccd2ca4de26fbe2bbcb55af3b8":{"kind":"REWORK","_number":1,"created":"2025-01-30 09:52:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/940432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/940432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/1"}}},"commit":{"parents":[{"commit":"2ead294761b1e418efdf294d2ade086c2e69dfc1","subject":"Merge \"Skip grenade job for doc update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2ead294761b1e418efdf294d2ade086c2e69dfc1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-30 09:52:39.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-30 09:52:52.000000000","tz":60},"subject":"deprecate eventlet monkey patch within oslo.log","message":"deprecate eventlet monkey patch within oslo.log\n\nChange-Id: I35797e0d03b69d77733a49c80b6e06c964fb48bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf240676f1a8cdccd2ca4de26fbe2bbcb55af3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bf240676f1a8cdccd2ca4de26fbe2bbcb55af3b8"}]},"branch":"refs/heads/master"},"f1158338dfa401111045f90c97c7f0a5535b23d7":{"kind":"REWORK","_number":2,"created":"2025-02-03 09:35:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/940432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/940432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/2"}}},"commit":{"parents":[{"commit":"2ead294761b1e418efdf294d2ade086c2e69dfc1","subject":"Merge \"Skip grenade job for doc update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2ead294761b1e418efdf294d2ade086c2e69dfc1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-30 09:52:39.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-02-03 09:35:41.000000000","tz":60},"subject":"deprecate eventlet monkey patch within oslo.log","message":"deprecate eventlet monkey patch within oslo.log\n\nChange-Id: I35797e0d03b69d77733a49c80b6e06c964fb48bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f1158338dfa401111045f90c97c7f0a5535b23d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f1158338dfa401111045f90c97c7f0a5535b23d7"}]},"branch":"refs/heads/master"},"e9730dacb2a3915d926863873895b1f28280ca5e":{"kind":"REWORK","_number":3,"created":"2025-02-06 15:10:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/940432/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/32/940432/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/32/940432/3"}}},"commit":{"parents":[{"commit":"2ead294761b1e418efdf294d2ade086c2e69dfc1","subject":"Merge \"Skip grenade job for doc update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/2ead294761b1e418efdf294d2ade086c2e69dfc1"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-01-30 09:52:39.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-06 15:10:14.000000000","tz":0},"subject":"deprecate eventlet monkey patch within oslo.log","message":"deprecate eventlet monkey patch within oslo.log\n\nChange-Id: I35797e0d03b69d77733a49c80b6e06c964fb48bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e9730dacb2a3915d926863873895b1f28280ca5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e9730dacb2a3915d926863873895b1f28280ca5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
