)]}'
{"id":"openstack%2Foslo.log~943512","triplet_id":"openstack%2Foslo.log~master~Iac1e5f98d9dd1dd423789cbe39589e93d62fb5e1","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2025-04-01 09:50:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1e5f98d9dd1dd423789cbe39589e93d62fb5e1","subject":"Use updated \u0027project\u0027 keyword for config docs","status":"MERGED","created":"2025-03-06 00:31:15.000000000","updated":"2025-04-01 09:51:30.000000000","submitted":"2025-04-01 09:50:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943512","meta_rev_id":"95ab1e08c702f8caaa364b3d63f2b8d1168e6737","_number":943512,"virtual_id_number":943512,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 09:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-04-01 09:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-03-06 10:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-04-01 09:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-06 00:39:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 10:37:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-04-01 09:28:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"95297fee1f178b3d33b60300a5d15fe098b96f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2025-03-06 00:31:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a4fd03a5f54996458eeb044b8b0a5c0fa09a6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 00:39:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4733f4e8d041608a86b4ad99530dc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d2eace4077419ab4815cf3c041a529 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83584652a8c24f809aa3cff285c5ba0f : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e317fadd3ddd3a53ff9dd9e850d79928a92c292e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-03-06 10:37:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40833f4da9957805efd2d82f76e99ec2598c4d63","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-01 09:28:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b94f1056ba20a8af6c5e53347bf92a7afd395d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:28:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ddbdf0adb7d905d99c94393355472667145267d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:50:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3cf1ec21eaa4aed9cbb222f687cf6c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5f39221d004170bdb881d2dfc52b73 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cad10dce1a74eb9a3174d9c7366d6ee : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"73c1ea044d7c4cd2e9cc9eacdd8ada32d0fb78c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:50:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95ab1e08c702f8caaa364b3d63f2b8d1168e6737","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:51:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd56f085b33467dbde1847973b29e7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defcb87e40e64878b59fd7d2c9674297 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36356598717d1f63e225980667fe2464e5e0c4d5","revisions":{"36356598717d1f63e225980667fe2464e5e0c4d5":{"kind":"REWORK","_number":1,"created":"2025-03-06 00:31:15.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/12/943512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/12/943512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/943512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/943512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/12/943512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/12/943512/1"}}},"commit":{"parents":[{"commit":"0c290b13a9c182b9149a3a03db9d9af28772bcb2","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0c290b13a9c182b9149a3a03db9d9af28772bcb2"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2025-03-06 00:27:42.000000000","tz":660},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2025-03-06 00:31:12.000000000","tz":660},"subject":"Use updated \u0027project\u0027 keyword for config docs","message":"Use updated \u0027project\u0027 keyword for config docs\n\nThe \u0027project\u0027 and \u0027project_name\u0027 keyword has been in used since Change\nId6d4b6eaa2b96ba5bd53e8b2dae5d9eea13b25d9\n\nAlso updated other minor docs\n\n[1] https://opendev.org/openstack/oslo.log/src/branch/stable/2024.2/oslo_log/_options.py#L181\n\nChange-Id: Iac1e5f98d9dd1dd423789cbe39589e93d62fb5e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/36356598717d1f63e225980667fe2464e5e0c4d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/36356598717d1f63e225980667fe2464e5e0c4d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
