)]}'
{"id":"openstack%2Foslo.log~947123","triplet_id":"openstack%2Foslo.log~master~Ic421ed988c41d2724a3a07e0b1136b0e731e5f8e","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-01 09:07:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-17 21:16:58.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-09-17 21:16:58.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-07-15 18:03:59.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"Ic421ed988c41d2724a3a07e0b1136b0e731e5f8e","subject":"Remove watch_log_file","status":"MERGED","created":"2025-04-14 13:59:44.000000000","updated":"2025-09-17 21:18:46.000000000","submitted":"2025-09-17 21:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947123","meta_rev_id":"87f17efcba82cb8745dc2437e3bc56e062f919d6","_number":947123,"virtual_id_number":947123,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-17 21:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-09-01 09:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-15 18:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-09-17 16:59:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 14:52:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-04-14 15:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-29 14:59:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-05-29 14:59:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"9d4bf8df084be96cf13b3cbaa3b9fc7f29da6956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-14 13:59:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8baaaaad6f73675aebaf895c9a98d6d8b23aa0dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-14 14:52:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40b0aafdefbeb74fae5701c5e6337d67e7ce3cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:39:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1749d31ed3dc47f6866cfcae90035ae3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e39ba3e813a44c0bbace2d6b95667c5 : SUCCESS in 1h 37m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd15fbf1753491892a394172ff5d62b : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f646dc5323f454ca6908ea0663662e8 : FAILURE in 1m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8b2038e4281494f95464d4008a96c89 : FAILURE in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbf764f26f174e5388b5af9934ed03dd : FAILURE in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c28a30bff1e4a4c9fd4915934ce1b8f : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341c2c34cc714a93ad339f5a42f04c70 : SUCCESS in 3m 48s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/a1e9b84bdd52425bbdfa6bb00931f10a : SUCCESS in 54m 55s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/04421eafc9cd4da6a14e1dd0457e276e : SUCCESS in 31m 27s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/8200cd867e38473db33004c9b8f309e3 : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"09ca703847a0d1e0335da64d775b40a2e675df4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-15 01:11:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f78ca3da9962b0a27c8e0f1cf616e587efdd951c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 02:41:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4be25936f134df086e19e8d944cbe7e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7f3a2871dc54930935613d9bfbcdeab : SUCCESS in 1h 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365646b89bab42649d45a199e19f0472 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99ae51d71b26470c9c2057ea6eb32131 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10db23fc1da44969880276dc923881a6 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36c8cd9ce56b458d89f8c0ff34395993 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a485c2532b34e4a8e1c745fc68702a9 : SUCCESS in 1m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6e9927babf4c2ea667ab09c7fb4007 : SUCCESS in 2m 26s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/e61b6043e951425d8341b54be3229864 : SUCCESS in 46m 47s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/1f3193312e294a2ab814f5b23250c680 : SUCCESS in 25m 05s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/81881604b20b43ad96572be2f620d266 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"eea2f130fe6f1c19e3176a30e8391843bfbd1c61","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-07-15 18:03:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21053b22b78c29bdb90676be45e4b5820cd1e2bf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-01 09:07:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2895947e7c0af73d18f7b00435bd46fbc111a597","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-17 16:59:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8aa8867ff8a834711c331ce1447d46c3e06ca80a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 16:59:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f62cd9b74e0d968f4cbdab2e1dc882355c8de862","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 21:16:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0de729bb7dc64e8c91a4ff7f31a7f487\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8659056ce79e4572872c4b15e477b3f6 : SUCCESS in 1h 37m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad30bf9b6c84d609a9d289ab992f00a : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88fd6e43902142c488c5acfd6ff49081 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a98d0fa02c2a4fb69334300e46d62aac : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f353cd1e3ef451cb0a032ea4ccfd770 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11fa2aa50de2443aac9dca439367b485 : SUCCESS in 3m 23s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/7e29c29647fa4435aed2ec30484dbb90 : SUCCESS in 26m 47s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/fd4ef63a61fb4f549cce00d78cfabea3 : SUCCESS in 3m 21s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/4853848d6d7947eab2f08ef362e70497 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8fe9e28d88e2f9294da85f284196a06a790bffa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 21:16:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"87f17efcba82cb8745dc2437e3bc56e062f919d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 21:18:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/651c3deb46c543d484f17991234064fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfdf33d002b4141967143445d4746eb : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe2aeb5670394a309b76c985557c9e3e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3e6dcaf2f2ec5cd259900ffd06728530ee2c57d","revisions":{"cd70842a0ea49e817d890de759140282b3b2bf00":{"kind":"REWORK","_number":1,"created":"2025-04-14 13:59:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/947123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/23/947123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/1"}}},"commit":{"parents":[{"commit":"9a58199588055a17c4df3e0e9130f7ca63af7e11","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9a58199588055a17c4df3e0e9130f7ca63af7e11"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 13:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 13:59:00.000000000","tz":540},"subject":"Remove watch_log_file","message":"Remove watch_log_file\n\nThe option was deprecated[1] and made unavailable[2] during 2024.2\ncycle.\n\n[1] 31603ac6dacd0b47ec9f02d99bfdd0a7b683119c\n[2] b977cb9becbeaaa67bb663bd8db5aa6846e495b9\n\nChange-Id: Ic421ed988c41d2724a3a07e0b1136b0e731e5f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cd70842a0ea49e817d890de759140282b3b2bf00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/cd70842a0ea49e817d890de759140282b3b2bf00"}]},"branch":"refs/heads/master"},"e3e6dcaf2f2ec5cd259900ffd06728530ee2c57d":{"kind":"REWORK","_number":2,"created":"2025-04-15 01:11:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/947123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/23/947123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/23/947123/2"}}},"commit":{"parents":[{"commit":"9a58199588055a17c4df3e0e9130f7ca63af7e11","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9a58199588055a17c4df3e0e9130f7ca63af7e11"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 13:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-15 01:10:51.000000000","tz":540},"subject":"Remove watch_log_file","message":"Remove watch_log_file\n\nThe option was deprecated[1] and made unavailable[2] during 2024.2\ncycle.\n\n[1] 31603ac6dacd0b47ec9f02d99bfdd0a7b683119c\n[2] b977cb9becbeaaa67bb663bd8db5aa6846e495b9\n\nChange-Id: Ic421ed988c41d2724a3a07e0b1136b0e731e5f8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e3e6dcaf2f2ec5cd259900ffd06728530ee2c57d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e3e6dcaf2f2ec5cd259900ffd06728530ee2c57d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
