)]}'
{"id":"openstack%2Foslo.log~961057","triplet_id":"openstack%2Foslo.log~master~I4b08406ee44a9de787afa7347a66b3529bebc9b8","project":"openstack/oslo.log","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-14 05:33:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4b08406ee44a9de787afa7347a66b3529bebc9b8","subject":"Move upper tips job from 3.12 to 3.13","status":"ABANDONED","created":"2025-09-14 03:56:55.000000000","updated":"2025-09-14 05:33:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5bba42f7dc55a37e41608d530103240ce31d894","_number":961057,"virtual_id_number":961057,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-14 05:32:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 05:32:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe4ca4fb563c25e2ebcb357ea35993a1925ed21c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 03:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"612fe5f3806cb90a3f4236817bb0a6e85f7fc911","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 04:00:03.000000000","message":"Topic set to py313","accounts_in_message":[],"_revision_number":1},{"id":"bc26db73a037a28e4578c5e133a9b5557d48b378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 05:32:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6222dce25ff460a8625ee8af592046b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e19171e09e62462799e2260e1a20475b : SUCCESS in 1h 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a96b77187c54f359264b06be9c04036 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/612c4f8a5eb44d63b4e53186f764e84d : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe86b6d77b4c4d089975f2cd15414be1 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81205bc8b9dc460eb9ca5cc1da84e3c5 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe34ed25eb347b8a34e876daf3f7f70 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53916afc98374ce58c06bff1fe032b88 : SUCCESS in 2m 41s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6001f8b7b4d84b7dac091742defd4d15 : SUCCESS in 58m 36s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ba0e3534f94b4c1c9ec2b9a9f58cbeaa : SUCCESS in 35m 22s\n- oslo.log-tox-py313-eventlet-patched https://zuul.opendev.org/t/openstack/build/985eedefc67e4053b391f25f6dc46479 : SUCCESS in 6m 10s\n- oslo.log-tox-py313-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/7f1b70bdfb2240efb6fd047d5f32c171 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a5bba42f7dc55a37e41608d530103240ce31d894","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 05:33:58.000000000","message":"Abandoned\n\nmaybe we should keep using py312 due to asyncio","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbba3db5f8e03bbbd6dee3cd75e0b3fd4d935ab4","revisions":{"bbba3db5f8e03bbbd6dee3cd75e0b3fd4d935ab4":{"kind":"REWORK","_number":1,"created":"2025-09-14 03:56:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/961057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/57/961057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/961057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/961057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/57/961057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/57/961057/1"}}},"commit":{"parents":[{"commit":"777be9527ab842ed32a4cddb7b3ca4757dcadd40","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/777be9527ab842ed32a4cddb7b3ca4757dcadd40"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 03:56:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 03:56:46.000000000","tz":540},"subject":"Move upper tips job from 3.12 to 3.13","message":"Move upper tips job from 3.12 to 3.13\n\n... because 3.13 will be the latest python version supported in\n2026.1 release.\n\nChange-Id: I4b08406ee44a9de787afa7347a66b3529bebc9b8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bbba3db5f8e03bbbd6dee3cd75e0b3fd4d935ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/bbba3db5f8e03bbbd6dee3cd75e0b3fd4d935ab4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
