)]}'
{"id":"openstack%2Foslo.log~969061","triplet_id":"openstack%2Foslo.log~master~I181516dae9243c41849a4b5399e9ef1e4a2fbaa9","project":"openstack/oslo.log","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-04 18:02:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I181516dae9243c41849a4b5399e9ef1e4a2fbaa9","subject":"Add ruff","status":"MERGED","created":"2025-12-01 13:44:07.000000000","updated":"2025-12-04 18:58:08.000000000","submitted":"2025-12-04 18:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969061-typing","meta_rev_id":"01469b0de6a7e24364548fc03f3a817fc07f3181","_number":969061,"virtual_id_number":969061,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 18:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-12-04 14:50:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-04 18:02:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 15:53:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 14:50:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b343ec3f5bdaecb1e5207a793c3134ad3e8d4a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 13:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dc0a6fbf306ac49596ab09423cdb15d51ebc14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 15:53:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16c62c89d7f14a82857341d101455980\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27973259b5434836b7154effcdcc229a : SUCCESS in 1m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/20180df4db234dc5af6ac2320f979132 : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c06579ce799248ee953eeddacb4b47ee : SUCCESS in 2h 04m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f237a9ca31b4d4ca832fe5238f7cb21 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01f96f5c98084087a26e6c69c9f3b55e : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4de8b528f97b42399c05966fb8a4ebb8 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486ba1895a5747e2b87d80cd7151d338 : FAILURE in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5fbfc0ac6044d8b892de967f8510e3e : SUCCESS in 2m 16s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/8b51354279fd46e1b7074bdc3bf6205e : SUCCESS in 29m 58s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0e6900e1d82f42feb70a32e60828a0b2 : SUCCESS in 17m 11s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/75e4a31aa8474eaeb6cf11822b2f4dbd : SUCCESS in 2m 44s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/af07cd26a05b4705a76d289295d80fd9 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"acbf340131efebbbcd413db93c0bc422d83f6c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-01 17:06:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"335536a0396fa0de5c6a97a596b85df7e824ed03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:10:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b848f539ed243aaa661b42f2fbcecd2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3d6b493a8a924bebae01f698a6452c85 : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/429cbd44c5744457a3cc26a88cb821cb : SUCCESS in 1m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7806e0bf6121445ea0a8b7972a702a1f : SUCCESS in 1h 58m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1370062764448de8171d47571c27b56 : SUCCESS in 1m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffdbcb73a15b4fc89903270294d23bc8 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2338facaeb54d1c98e75a620d8ea6db : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14414f4a6fc3410082afd61f9ea0b181 : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77766a43a6cb4695af45451254567c9e : SUCCESS in 3m 11s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/f97491dca49f455097374f69004b5723 : SUCCESS in 55m 18s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/6aebb78fe376478e934594d5293e86d2 : SUCCESS in 35m 32s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/8f762ba6287e485587ed817891850ccd : SUCCESS in 2m 15s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/c33a92ea804145ba8e81916ab4943f1d : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ad6a21f4a4ebe642dc2433753d4cd3971a97a6a4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-04 14:50:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ef43d74da18516ebf7000cc1e7b8c4bfd2b5fd27","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-04 18:02:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8987e47bff613a9837a7ae5a3741c9f821edc87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:03:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7420b701908cb9c164f3da3ca41470cd000f4d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:57:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b931e1b067f64dd69dc679e4970265a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3bf20cde86cd45e4b2263ba0249cf354 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21095855ce37413cbb50b4d2a0888eb6 : SUCCESS in 2m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/008014095f7d40a5a325645db48b1874 : SUCCESS in 53m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41d295c5bbb412b8c1b08a57b5d2ceb : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bf53adf38cc484eab6667dd5e42333d : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8fc43994ceb4ea0942efcb95df6625d : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f14e8de85540eaa8bc1fca6872c61d : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01d4c9cd6de4460b2bf0d70b4faa247 : SUCCESS in 3m 27s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ff5829fc9f15481ea0b8078a960b2841 : SUCCESS in 40m 42s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/56ac737f65fd49d2afc091c039ac2bf3 : SUCCESS in 2m 04s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/333db1df4fc64217904036286464daf9 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"32770c6bb49c1c12aa40a97cccd5f6149de811f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:57:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"01469b0de6a7e24364548fc03f3a817fc07f3181","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:58:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d4f53ce704495baf295c43dadd7d57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce8332587134c48b262966049301014 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab3bbb2a87b4721890b4a7a3473df18 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1279f7b9854840fc44fa0b01f50acc23e09e6f7","revisions":{"802d308a2c2071989d10e9f2d7fd6f15ef607390":{"kind":"REWORK","_number":1,"created":"2025-12-01 13:44:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/969061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/969061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/1"}}},"commit":{"parents":[{"commit":"df331c63b70875eb1d18a58443c5e520ca4ed88d","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/df331c63b70875eb1d18a58443c5e520ca4ed88d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-09-30 16:19:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-12-01 13:43:00.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I181516dae9243c41849a4b5399e9ef1e4a2fbaa9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/802d308a2c2071989d10e9f2d7fd6f15ef607390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/802d308a2c2071989d10e9f2d7fd6f15ef607390"}]},"branch":"refs/heads/master"},"a1279f7b9854840fc44fa0b01f50acc23e09e6f7":{"kind":"REWORK","_number":2,"created":"2025-12-01 17:06:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/969061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/969061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/969061/2"}}},"commit":{"parents":[{"commit":"df331c63b70875eb1d18a58443c5e520ca4ed88d","subject":"Remove reference to tag framework","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/df331c63b70875eb1d18a58443c5e520ca4ed88d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-09-30 16:19:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-12-01 16:51:52.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I181516dae9243c41849a4b5399e9ef1e4a2fbaa9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1279f7b9854840fc44fa0b01f50acc23e09e6f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1279f7b9854840fc44fa0b01f50acc23e09e6f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
