)]}'
{"id":"openstack%2Foslo.log~970584","triplet_id":"openstack%2Foslo.log~master~I53fa4f46ebfd015803c867410fef662eb68bf340","project":"openstack/oslo.log","branch":"master","topic":"ruff-log-checks","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-11 13:45:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53fa4f46ebfd015803c867410fef662eb68bf340","subject":"Enable logging related ruff checks","status":"MERGED","created":"2025-12-11 08:01:47.000000000","updated":"2025-12-11 13:46:16.000000000","submitted":"2025-12-11 13:45:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970584-ruff-log-checks","meta_rev_id":"7f2ee02c684d7ad5b02004bb9d07956f91ed160a","_number":970584,"virtual_id_number":970584,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-11 13:45:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-11 12:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-11 12:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-11 09:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-11 12:30:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b7fdcbf3887c394d8b35a0feb532cdab32e16832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-11 08:01:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46ec59e50391fa40daab7b92965f6770170f5807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 09:49:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e8147fd118a485f8d6e7f99a87766a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9b916b8116e43bf97b06b9e5aa2b9c6 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf54e2bdd6544f9e9d0daecfc59eb4e5 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd5669a6b0b24de192490f353a3f30a0 : SUCCESS in 1h 42m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e93b2ee19f48e9a7f3b21fd473b919 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a65d654bac1740a8ab3deaa11b23e5a2 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1659d323ecc4a4a93de3428e24ef086 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53977d983d2148ddbc5f398362826b95 : SUCCESS in 2m 34s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/affab2d8c381413080d6ece7bb2cf82b : SUCCESS in 1h 02m 08s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/9e3f7063a9e546e692ee6f083c0f189d : SUCCESS in 16m 59s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/92cce5486080428abd6e869c01547dbe : SUCCESS in 2m 12s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/57328a6e26044d59a5b19f5bd8ec2225 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6564a9773ecd1afa10e6b6d158014e42692bbe59","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-11 12:30:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7a4e9ae7865658615053a03cccd2700bfd6b63b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:31:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7651e43f74b8361a8d035b3b2815abf587c540af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:45:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/864653f0402b47f7ada2172982389f58\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a97915f13824d48bfd979386c8a99a5 : SUCCESS in 1m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0abc2fcd17e74eeb98f5c87290e4f525 : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce255e2ef37345f38816367a8ee9341c : SUCCESS in 1h 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1eed226c4f4214897ba3ab318a2f1f : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9e5fd99a1f8445f9035f707c25ea976 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02fabc63629449e9810fe5d104692a41 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdfad44cc1a4b718d7c2434ca09dcb4 : SUCCESS in 2m 56s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d7ee8aa9911a4136bf59bfed99ef505a : SUCCESS in 18m 50s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/5c7ec7164e0142bb936a8811ab5aeeb9 : SUCCESS in 4m 32s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/08110a23fee140db8cb1658a6f1ed171 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"350019033d69315083bf22f62cdd81964d71fb83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:45:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f2ee02c684d7ad5b02004bb9d07956f91ed160a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:46:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693c39be6b134820b13e8d0e25c7fb78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd5b3e60f6449f7aeab28a4f3e8624e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"310f6c889532d75836d69e5ef6990a8b9292d210","revisions":{"310f6c889532d75836d69e5ef6990a8b9292d210":{"kind":"REWORK","_number":1,"created":"2025-12-11 08:01:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/970584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/84/970584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/970584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/970584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/84/970584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/84/970584/1"}}},"commit":{"parents":[{"commit":"a1279f7b9854840fc44fa0b01f50acc23e09e6f7","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/a1279f7b9854840fc44fa0b01f50acc23e09e6f7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 08:01:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-11 08:01:37.000000000","tz":540},"subject":"Enable logging related ruff checks","message":"Enable logging related ruff checks\n\n... to extend the enforced format rules related to logging.\n\nChange-Id: I53fa4f46ebfd015803c867410fef662eb68bf340\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/310f6c889532d75836d69e5ef6990a8b9292d210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/310f6c889532d75836d69e5ef6990a8b9292d210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
