)]}'
{"id":"openstack%2Foslo.log~971338","triplet_id":"openstack%2Foslo.log~master~I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7","project":"openstack/oslo.log","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-02-17 18:54:19.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-17 17:04:29.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7","subject":"versionutils: Add support for post-Z releases","status":"MERGED","created":"2025-12-17 15:35:00.000000000","updated":"2026-02-17 18:55:11.000000000","submitted":"2026-02-17 18:54:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971338","meta_rev_id":"a2a7cad5e86bb578a9e4c3db04715a058c3f4af4","_number":971338,"virtual_id_number":971338,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 18:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-02-17 17:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-02-17 17:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 17:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-05 10:49:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"456e534014e7a50967345095e1ac3d040d0fd55c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-17 15:35:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a254d817f53a9d8c71b18c499d27de2189bfb71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:21:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd3f6e5c4a094576b0f3139ee6f673c3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3982f53bb97b419bb2401f0352812e1a : SUCCESS in 1h 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0569d861532f428ab69a18328e1f8554 : FAILURE in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5e19bae6df046b0aea7c315e2d6d966 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0563ca448ef6429f8b2cbb1863278c86 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfb6ae0307c40e9b743bbaf59f3c50d : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60e15afe2a064ad6b75fa4d53a114eb3 : SUCCESS in 4m 08s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6b2c5d2003ef4fd38ab446ecee609103 : SUCCESS in 1h 02m 23s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d65590e947684ce294eca4be0a27c599 : SUCCESS in 29m 08s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/c9e905621e3343c29c3bc8526786c62f : SUCCESS in 3m 01s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/7eb1912dcd4a4d7aa89c5663ec95182f : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c5ff95fac3f05ae9ae475987b4add10301c2439c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-17 17:25:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"12729e99869711950208340a4fee0bf95f22b111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 18:22:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6fcadc1e7214a33aca6f837bac2cc36\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03a2daeef65f4028805e83f0f5f0e610 : FAILURE in 55m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0b9e5100b347608c1c6a3b741815c8 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98a6c24f4a394e07b509003b33169824 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/206ecc8f522342dbb3e46d312cb221a0 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c529ba04e74113b91afbe552396f22 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c27a20086c034bb7a3429bb840100c0c : SUCCESS in 2m 57s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/05bdb1df7e224d23b7a76ee288619e69 : SUCCESS in 37m 05s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/2fb88e2b2ff74b3bb52586f780358d7c : SUCCESS in 36m 56s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/73ed6055629d45fda966ce513d535482 : SUCCESS in 1m 57s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/d2c111513f3548df847c852a4efc12bf : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"99cebfd453a79f2abfa5f93e1c4b58fc78c7437a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-18 09:36:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a787596fc657e4c0a69e1f6f3d33da695c3377e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 10:47:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b5c1d83dbc486fab86016635aa4464\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41134302616c40d0b4abec3335286104 : SUCCESS in 56m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be561e7a364412f95107696d5d063c4 : SUCCESS in 1m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03dc6ebcddd84d82b89cf381c5e74e54 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2bc388e26ca41c29ec14bbfe255953f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a49d43ddc84f838756838142d99fda : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4d070391b8f4f42ab218d3338f63189 : SUCCESS in 2m 23s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/252ca109ea3c48d4851482f1a95ab8e2 : SUCCESS in 1h 05m 30s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/53d45cbcdcf64c5687beb31ecf95e83d : SUCCESS in 39m 23s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/0ccafe04b23347e1ab876f7ec5924865 : SUCCESS in 2m 43s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/7068d669a8ed44c9beb1392e20248a5c : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"298adce44827fa2b7ccaa8bf721205112becc993","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 10:49:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ad2543771c9a0f2185e24c48421aaaee75c925bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 12:05:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d39ab1c55ce5661c32ddc540b82a5c1241974f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 12:07:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2146d2a5a8f253202c0643c4b813710cfaa7e770","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 12:08:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cbec46bd8e51fbb9f0a5f7b11a6d66cb0b12a3dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 13:30:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89009e11b8d4b0fae32204d7eb51c3c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce9480a42e2a402bb2e0031cecfee04b : SUCCESS in 55m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26d37d58ee8e4b148641a871dbc4a192 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdd837326a324f609021e43d6027f3d6 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a705999196504390a86386396e488d11 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4757f42f5bdc41619290778aa6c0ec04 : SUCCESS in 1m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc75dadfa28e46a787cceb2808696749 : SUCCESS in 3m 00s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/7162377db9f84c908fea45b31c6886ff : SUCCESS in 1h 02m 19s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5f6cbc75e7ee47129c0434a9b807a4f8 : SUCCESS in 17m 35s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/fecb194b6d4048b6aa3d087c22d82e06 : SUCCESS in 2m 56s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/4ca8e3d8c2e54d5e9910c09b8313ad41 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":4},{"id":"91ffb2398787ef6955855f5775099dab74b1fb1d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 13:42:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea21cbe49d69adf1d7047cbe3a57b9775ac05363","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 14:26:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67bfee76c0b8ce3a483b889e2444913ca70ca082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 14:27:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"27bdf06f0868b14501fac4a9eb0ffa090494b628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 14:27:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f351f2b2aff6f462c0f33ffb169e09531febb819","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 14:29:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b933b1918e8768dbe1837e5ca922800c8683712b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 14:46:10.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f3e4a0b6e6b8ddb9cf4f9ea2635258fc2f475dd1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 15:23:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52569ff8b6fd4ba54c2b06bb6ebce2d62cedef72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:08:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/358ccdae8f664a4abe63b1211021ef6e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d12f07fe0c344b3a94b0e649dad68acf : SUCCESS in 1h 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647b22f7b280423f9468dba66fa19f3a : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b1084ff565d400fa7d64f7b6aabe937 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0322feb08e3a47b8ba198a6ae38d72c9 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b34fbbee564ac3a9ff399ab010774c : SUCCESS in 1m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01a152194a74fc1a4bf4dff5a92a5de : SUCCESS in 1m 46s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/78197424153a4a33a312931b2cb31ff4 : SUCCESS in 1h 02m 29s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/ca41930a83464960b536983a81c842ea : SUCCESS in 40m 31s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/3bb3a271529b4b2e93f86febb8126a09 : SUCCESS in 3m 06s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/826e661e06c04137a1b32650f695db49 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"bc191e8d7b99d46a82f541d42e0ca46c7414bc95","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 18:54:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"331d2299e424d82891e54fbdc0275d97143101ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 05:33:05.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c42827c92616996beb4ff68796759e57eb80321f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 05:33:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51ddede4392572c39f0b82f73f6c2792ae447512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 06:49:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae9efb323b914c2e99f9b780e4787e25\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa5020fbaf354d54aacc663126d5b79e : SUCCESS in 1h 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178d7fe9f70847deb2a8f55cd57d3fcc : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e4fe243619d4cc2b2b6c015f151f917 : SUCCESS in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55ba0be1bfce4badb2369013fd1578bd : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e1cf92da9e4fddaa3b6ae2602ddf30 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3772cb6c4f594b8bb17b639e5025c067 : SUCCESS in 2m 29s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/d2c064f6e07947c3aa1032aceebb9e70 : SUCCESS in 58m 06s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0963bb0d27e14fcf975e9fac4f7bf527 : SUCCESS in 16m 59s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/def87953a3b246f6bce731244a457ce5 : SUCCESS in 2m 48s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/293f2cbd3f6341eaa0ff9125f6b258ee : SUCCESS in 2m 04s\n\nWarning:\n  Comments left for invalid file oslo_log/formatters.py","accounts_in_message":[],"_revision_number":7},{"id":"449b1e70235162d0f5759ac60e84897e692bea01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 08:12:46.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a0a416c81cfc1d98df99bdf82d2f432adb1c516a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:59:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bafe27ec35c4a1ab7af7e2968587bd5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efa8e3ef4c954652a3afff2dbade6d1c : SUCCESS in 1h 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c296cfbeee4c77858b1f2f6b5d65c8 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afa64574e666465da3ab7ae3ecf64bc1 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a9770956c8a48799d2db7d46c1b8180 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b585bc7d2ddf4f00836c75c7f0a95645 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f77e6076580c49f1af0047550e1600d8 : SUCCESS in 3m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/509015c72aed49269fb23d41288676aa : SUCCESS in 59m 34s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/0b199a7f8de94a60b122caaa4feb131d : SUCCESS in 23m 55s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/9c1a05c73709445780b0729ee1103d82 : SUCCESS in 4m 34s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/64296be47ec94f40a4dc56a1577c9f49 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":8},{"id":"32de7afa1cdeadb2e08cdba499459f4a7d72a51b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-17 17:04:29.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b5d7fa7b6691be73c98ad27b7f5b7e945ab0b8e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 17:05:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"db5272be29e0302723d913349032ec2cc413f394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:54:19.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b979cba1b92640989ba8dc76710aa5a8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e796440d658a49bd9aa246506f93a57f : SUCCESS in 1h 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99afbd5bcb0e4fe7a96ebac221466011 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c3c913ad6284f6e98db4ce5eadaf160 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67dac9806bca43529fa391557b0bc32b : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4559120561ce4b7bbbc6c3f4da3e6866 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53e22e3ea64c4c3dbd0bfd75520496bf : SUCCESS in 3m 51s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/cf7f5cc9b1344776b601b9a2b6cdd4c6 : SUCCESS in 18m 58s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/244a862ac42b4dccbe32b25c47e0f5bf : SUCCESS in 3m 52s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/21d2e9e0654d4b7e9a875bf31e98ec4f : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":8},{"id":"f9d738a467c9dff45123ec4566874e79c18455a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:54:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a2a7cad5e86bb578a9e4c3db04715a058c3f4af4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:55:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e40dee9ac94deabbb0e1459d35f34c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dad4ad94654ea0b18d8d06f3adf02c : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9911620a73684bf082729f3cdc1f5629 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d0f0bf3ee5d5676b99b4ad7656089808c2aeb35b","revisions":{"f182e382a62c3f24bbf6e0853a42c3c5f98c381b":{"kind":"REWORK","_number":1,"created":"2025-12-17 15:35:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/1"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:34:49.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f182e382a62c3f24bbf6e0853a42c3c5f98c381b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/f182e382a62c3f24bbf6e0853a42c3c5f98c381b"}]},"branch":"refs/heads/master"},"e5b1b000a9ed330b843d1e668585de34668c8de5":{"kind":"REWORK","_number":2,"created":"2025-12-17 17:25:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/2"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 17:25:09.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e5b1b000a9ed330b843d1e668585de34668c8de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/e5b1b000a9ed330b843d1e668585de34668c8de5"}]},"branch":"refs/heads/master"},"5881303a7e1040929500311d473b28f30f372b90":{"kind":"REWORK","_number":3,"created":"2026-01-05 12:05:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/3"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-05 12:05:10.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5881303a7e1040929500311d473b28f30f372b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/5881303a7e1040929500311d473b28f30f372b90"}]},"branch":"refs/heads/master"},"9f00aecc01ab5c7a1f0629661416247988c9005c":{"kind":"REWORK","_number":4,"created":"2026-01-05 12:07:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/4"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-05 12:07:44.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9f00aecc01ab5c7a1f0629661416247988c9005c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/9f00aecc01ab5c7a1f0629661416247988c9005c"}]},"branch":"refs/heads/master"},"145a06b0188633676d831fe58e94d4b1c92967f2":{"kind":"REWORK","_number":5,"created":"2026-01-05 14:27:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/5"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-05 14:27:06.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/145a06b0188633676d831fe58e94d4b1c92967f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/145a06b0188633676d831fe58e94d4b1c92967f2"}]},"branch":"refs/heads/master"},"3498e144a80caa2f26ceae9a8b67f52c636a94ef":{"kind":"REWORK","_number":6,"created":"2026-01-05 14:27:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/6"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-05 14:27:40.000000000","tz":540},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3498e144a80caa2f26ceae9a8b67f52c636a94ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/3498e144a80caa2f26ceae9a8b67f52c636a94ef"}]},"branch":"refs/heads/master"},"452e458e8f448a4e0cbda1f020d2d771627f17ac":{"kind":"REWORK","_number":7,"created":"2026-02-17 05:33:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/7"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-17 05:33:02.000000000","tz":0},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/452e458e8f448a4e0cbda1f020d2d771627f17ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/452e458e8f448a4e0cbda1f020d2d771627f17ac"}]},"branch":"refs/heads/master"},"d0f0bf3ee5d5676b99b4ad7656089808c2aeb35b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-02-17 08:12:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/971338/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/38/971338/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/38/971338/8"}}},"commit":{"parents":[{"commit":"719416866f45e3b194de5bec0ec166803df05ca2","subject":"Fix mypy error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/719416866f45e3b194de5bec0ec166803df05ca2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-17 15:23:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-17 08:12:46.000000000","tz":0},"subject":"versionutils: Add support for post-Z releases","message":"versionutils: Add support for post-Z releases\n\nOpenStack has already reached its Z release. Add handling for post-Z\nreleases such as 2023.1 (Antelope), so that consumers can use\nthe common definition instead of hard-coding release names.\n\nChange-Id: I7a0d907c50a6ab60f3c71c4f74dd5c0774dbdeb7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d0f0bf3ee5d5676b99b4ad7656089808c2aeb35b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d0f0bf3ee5d5676b99b4ad7656089808c2aeb35b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
