)]}'
{"id":"openstack%2Foslo.log~972176","triplet_id":"openstack%2Foslo.log~master~I18d20743fda1d52408cccb445dc23ad962b248f0","project":"openstack/oslo.log","branch":"master","topic":"typing","hashtags":[],"change_id":"I18d20743fda1d52408cccb445dc23ad962b248f0","subject":"Add typing classifier","status":"MERGED","created":"2026-01-05 15:01:10.000000000","updated":"2026-01-05 18:55:23.000000000","submitted":"2026-01-05 18:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972176-typing","meta_rev_id":"2d19970b2f252376fed03b62c339c0c5529286eb","_number":972176,"virtual_id_number":972176,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-05 18:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-05 16:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-05 16:39:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 17:03:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b98dea16286d380210d4c2d3354be172828ba07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 15:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc230e424b1c64c299bc1ccf1d01d218d2829817","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 16:39:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5e379d4932a6b63a8f9d8f8b25d06a1af5f072f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:03:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba977760feff4054b81654b3f9d7d97d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a708c9ec58c4c2bbf55e1dfcaf7290b : SUCCESS in 5m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/46898179a24849eea03703d0e6c75130 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02582924688b4bb4951606a1c3b5c617 : SUCCESS in 1h 43m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c9a17600d446708740eefa0e0179eb : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccaf5926f8c24509a1c5595a006e9361 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73ff9612b1ad4eb489b41baa77efaa8f : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26519a5b82a4546bcf21a313369cc17 : SUCCESS in 2m 42s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/57184caaec1249348a5dafc431ef28dc : SUCCESS in 28m 52s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/92afdfb511f6437d86e716fb44884cce : SUCCESS in 34m 21s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/508045f4beee47b5a43e1850ae0f0173 : SUCCESS in 2m 40s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/d575d7bbcaac42ce83fac693fff56185 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bfaddb623a6b9a270da6a34ade990cf4aca23a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1700805e672fe246ff26adc5dab7dc16b683cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:54:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c850f5775bb246f1ac03ace3ca50be32\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f876096cec3407d874ef923771f88ea : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02c6b9f545e5418e8fa57fa32fbe6e23 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/767f80b9bcaf4ef4a230aaa5b4a351a9 : SUCCESS in 1h 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e9a32cbaeb476eb353b5143e0afb8f : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f12546ef4ee044709134ea96cb96fbaa : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b07435bc10104ce89756b04d50d74de5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fbd923df1146ff81f7e9f752d63584 : SUCCESS in 1m 57s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5b727d79eec84a0eb26ae0f923e65f8b : SUCCESS in 19m 49s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/a38ad778d11d40129367f116aa4dff49 : SUCCESS in 4m 48s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/d67b95c31c294adc82837e5d82c91ddd : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cffe1985628a1ff588b67f768dce26bdaffc6150","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d19970b2f252376fed03b62c339c0c5529286eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:55:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409d5d38aec444d1ae0172593ec6c21e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495d0b481e594a20b3b7836d3947bbbe : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"930cfdf414d4ea4a36188401415e20ca3bfd8d4a","revisions":{"930cfdf414d4ea4a36188401415e20ca3bfd8d4a":{"kind":"REWORK","_number":1,"created":"2026-01-05 15:01:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/76/972176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/76/972176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/972176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/972176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/76/972176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/76/972176/1"}}},"commit":{"parents":[{"commit":"52ac599aa48fe5dc7c295ded8e7a87195e3e4c03","subject":"Merge \"Enable logging related ruff checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/52ac599aa48fe5dc7c295ded8e7a87195e3e4c03"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-05 15:01:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-05 15:01:05.000000000","tz":0},"subject":"Add typing classifier","message":"Add typing classifier\n\nThis is parsed by PyPI.\n\nChange-Id: I18d20743fda1d52408cccb445dc23ad962b248f0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/930cfdf414d4ea4a36188401415e20ca3bfd8d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/930cfdf414d4ea4a36188401415e20ca3bfd8d4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
