)]}'
{"id":"openstack%2Foslo.log~979299","triplet_id":"openstack%2Foslo.log~master~Ied6aeb05678991aacbba2d7347a4b034ff5d2c59","project":"openstack/oslo.log","branch":"master","hashtags":[],"change_id":"Ied6aeb05678991aacbba2d7347a4b034ff5d2c59","subject":"Skip grenade job and devstack job for test code update","status":"MERGED","created":"2026-03-06 16:04:23.000000000","updated":"2026-03-07 00:09:52.000000000","submitted":"2026-03-07 00:07:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979299","meta_rev_id":"0bd69e3a4da16c725840cb0e8609a9d56b39d3d3","_number":979299,"virtual_id_number":979299,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-07 00:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-03-06 22:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-03-06 22:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-06 17:53:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1100a7ebc9becf6dfa2e5a3ec2ba1cb81f336ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-06 16:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"859a9deff8853c1d6dceab7ebbe585a21a2ec07e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-06 16:04:36.000000000","message":"Topic gate-fix removed","accounts_in_message":[],"_revision_number":1},{"id":"2cebf93093462b869b011250607ed59b921e85aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 17:53:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4e9560f2374a2e8c66e61f82fdb125\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9580ba9fbc5143febe30c1dbe88ea713 : SUCCESS in 1h 41m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f498fcfc7a7d4712ae29ae21ce0d0758 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/955e24a3b3674b3b9a4d7e34cbd6210f : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4e411bfb17444f3931c5bc1e0ec2dec : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f95c59a30d634c3b97a4bf5a9c7b561b : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f528d31c744c4395a303dd36172a77 : SUCCESS in 5m 49s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/3f7cbdb5fa4c49edbc65ecddc5607f2b : SUCCESS in 1h 04m 54s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/4f02031e268e43889da5fc2c3c7fd9d3 : SUCCESS in 29m 22s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/02d3b0d8188a43448488952b3810df4a : SUCCESS in 3m 12s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/9cbbdb70eaa747858450d25e749f5199 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6fd47160b57d79315a7222cd60e484bf4a1ddf4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-06 22:05:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c235b9db487663634f3607bbb078243d12e3035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 22:07:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4162cf7ac8d617f403779055283c45335bb34fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 00:07:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44e2640852e47aa8cfb7b38572ec1c5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac353e9f41084b0782e6415273a0c1c6 : SUCCESS in 1h 55m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ef94349c144d01b098f8d64f2571da : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa0926076a8849d086b21ea1640c45b1 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a92b31e34da54a788e935fa641637467 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9905403925452c8443315773f33497 : SUCCESS in 3m 04s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a8b730f7c9f74ae898924aec51a761fd : SUCCESS in 35m 03s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/2d02c955e5d84e96be1490dc9753ce71 : SUCCESS in 3m 35s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/c20db9198f0c40f08fb20949e4842a0a : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f2dd666939daa31addce63a41a765b89fa43ebe3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 00:07:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bd69e3a4da16c725840cb0e8609a9d56b39d3d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 00:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17bf9e5c526f47e09fee479cc5969d58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc7e148b7af4c099cd3206ad2989a62 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6601232038de4672ef1e1c2879ae94c353779d4","revisions":{"d6601232038de4672ef1e1c2879ae94c353779d4":{"kind":"REWORK","_number":1,"created":"2026-03-06 16:04:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/979299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/99/979299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/99/979299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/99/979299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/99/979299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/99/979299/1"}}},"commit":{"parents":[{"commit":"d333016f740ec007c3502c17f6fc1599a54da1ae","subject":"Fix new mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d333016f740ec007c3502c17f6fc1599a54da1ae"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-06 16:03:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-06 16:03:55.000000000","tz":540},"subject":"Skip grenade job and devstack job for test code update","message":"Skip grenade job and devstack job for test code update\n\nChange-Id: Ied6aeb05678991aacbba2d7347a4b034ff5d2c59\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d6601232038de4672ef1e1c2879ae94c353779d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d6601232038de4672ef1e1c2879ae94c353779d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
