)]}'
{"id":"openstack%2Foslo.log~979561","triplet_id":"openstack%2Foslo.log~master~If325090ebf170b045dd797890e2e4c727002864d","project":"openstack/oslo.log","branch":"master","topic":"linter-sync","hashtags":[],"change_id":"If325090ebf170b045dd797890e2e4c727002864d","subject":"ruff: Configure hacking as external linter","status":"MERGED","created":"2026-03-09 12:19:13.000000000","updated":"2026-03-09 15:52:27.000000000","submitted":"2026-03-09 15:50:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979561-linter-sync","meta_rev_id":"47e1e7e13c4d72af7338e9f792ebce296f2f1ad4","_number":979561,"virtual_id_number":979561,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 15:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-09 12:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-09 12:48:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:00:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"defb0238714be8f286538bf3ac68de34999601f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 12:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"632842e2572084d23da32b569c4f5f40f9a0238b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 12:48:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0c1803da5a7cd075829cbe7b364de1667bbd93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c73e0538d6e247bbb66fe6a0a9661945\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2756558db6744066ab468ca7d7bfffc5 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f20137626a5b45a4a8437aa96fa33705 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/feeb1b0a180e4e958633c1a62e30d26c : SUCCESS in 1h 38m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28abc6255a04a9b9e0db921d3a79670 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/984ed0a392814ebf8b312be863c8f715 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ab6aa851e88435f9f3cbb7b13daa5a8 : SUCCESS in 2m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ca8191927624bf6b445d1d426540c61 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb7098627cd426caa64c4d0f502d209 : SUCCESS in 4m 51s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/ee3d0d0c73b648c3944435e75faaa78e : SUCCESS in 1h 07m 20s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/d44de84f58ee4049bfba555d1fddbe26 : SUCCESS in 29m 03s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/0b1acb7dbee84cf9afb9db23dfa13efa : SUCCESS in 2m 13s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/fdc880ce41854bf6adf216c5954cc8a6 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"995b821ccdaa5bd85be43d43f695d656e0f10809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:00:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7abf0fa1e8901700709f773f39d0d519c1211b08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:50:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f320583d56f74c558bb7002466de0d8b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2a3f462072644fd08318b74e7e7e68dc : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/65f7d4aaf8884c8096b8995148e939c7 : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f669eb6b5b2b4d8e9d6cb7c9317d1871 : SUCCESS in 1h 45m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3babef073dee41758f71a60848b54b19 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11727f2d327841cb876d6ea818c0ed45 : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fe18cba169d47a199d1eaa9bb146ada : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088c0e10276c40948b04387255e5c294 : SUCCESS in 3m 33s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/23da3e35d0734ddcbd8756bf09d807e5 : SUCCESS in 37m 30s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/22e874d11904464ca330aed66ffd2613 : SUCCESS in 2m 57s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/835d14f9f2804b0cb479f015a8822d5d : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e87603e74082eb2c795b6bbe9d6a72da23e53606","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:50:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47e1e7e13c4d72af7338e9f792ebce296f2f1ad4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e9b59ecd454ce9903555abe39dc795\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f966b9b7e7ac483dbce3db62ad0008af : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2818f131f4365aee3daa092bbae22230c8974b7","revisions":{"b2818f131f4365aee3daa092bbae22230c8974b7":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:19:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/979561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/979561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/979561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/979561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/979561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/979561/1"}}},"commit":{"parents":[{"commit":"825406d5d114dc7651f4edf9e9f9a0faaa5e9984","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/825406d5d114dc7651f4edf9e9f9a0faaa5e9984"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 12:17:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 12:18:15.000000000","tz":0},"subject":"ruff: Configure hacking as external linter","message":"ruff: Configure hacking as external linter\n\nChange-Id: If325090ebf170b045dd797890e2e4c727002864d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b2818f131f4365aee3daa092bbae22230c8974b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b2818f131f4365aee3daa092bbae22230c8974b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
