)]}'
{"id":"openstack%2Foslo.log~979562","triplet_id":"openstack%2Foslo.log~master~I88beb168eb5a01d5b696cd0da6ca696cdc3149ea","project":"openstack/oslo.log","branch":"master","topic":"linter-sync","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-04 12:07:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88beb168eb5a01d5b696cd0da6ca696cdc3149ea","subject":"Update packaging configuration","status":"MERGED","created":"2026-03-09 12:19:13.000000000","updated":"2026-05-04 12:08:31.000000000","submitted":"2026-05-04 12:07:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979562-linter-sync","meta_rev_id":"62959840232166e4ccf9b696b6b472b9ab2858d6","_number":979562,"virtual_id_number":979562,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 12:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-02 12:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-05-04 08:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-02 12:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-05-04 08:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:17:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-04 08:23:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8abad51dfbcad4518df208f6cc62f31080ec8e1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 12:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a19ab3e8079fe03ec02e7ce26c7548fc8e031cc0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:13:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"35815504bce79b82957f4107b9184c81d99df8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:17:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bc4b716d725476d898d472490169672\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e41ce8855c154752bed3be7e0c7d10b6 : FAILURE in 2m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/27ca77ab999d4b8bb8bd88c4e6be3a0e : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47108b21f36f4bf4ad4c04038684112a : SUCCESS in 1h 52m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4152648dad641658efd9f99bed90416 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/865de4e2b4e9407eb850b909aad7350b : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e58c07f4dadc4c65a521cca7145e4662 : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/130d6420bf714edea6e0455b19c70547 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66b1bd5ec4b4f1da5ea3f04f3d3918a : SUCCESS in 2m 57s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/4438bb1cbe0043ac971daff8a94bb2db : SUCCESS in 55m 32s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/de82fd5fa1e846f79aab71f325f0c3ae : SUCCESS in 35m 38s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/0f49178f62874566b44f5f52eb12f364 : SUCCESS in 3m 46s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/5bf2bdc3fab64dbeafa867c8d0bfa872 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"dd5ce231854c9f77596d77902bcd5451e843391c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:28:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4df828bf9c0bb786cdd9bbb95f995d1acdc62a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:13:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df93e0d23a3c42759eb29abda373ea6b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57d031834f6547a484512ec68d751a2a : SUCCESS in 2m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8189d1e0d50644808669aaf3b99090e0 : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fed63ac4052f420995d62eeb4cdf45ce : SUCCESS in 1h 39m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646fc1c753394e69afdbbf51e1fc89c8 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f66564cf554648fba2fe202692af46e2 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/721d1a07477c47d49fb15407771863f3 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b5e89aa02c04cb8a6da09ce417eefb8 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1d3cb111014ce3aff690f5c2b419d0 : SUCCESS in 3m 06s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/6d407889cd8a4f1fb06a5b1118907b0f : SUCCESS in 44m 58s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/e15a761297ad4093af3920bdb7f8fa42 : SUCCESS in 36m 37s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/de82d862b41746ad8ca96670ab563e34 : SUCCESS in 4m 50s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/41b289f0b185401385b2b8450e84e9f9 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"525100ef94c0dc3f151514c221d3f9f786e54223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:13:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c15dd940bb618a72038ff878cf67707f059ebce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 16:04:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02d336f3433b461293cdb38ff50dff99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f5c4ad7a23144bc48634857314a6b689 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b21cc10e8886465c8972acd3d8829786 : SUCCESS in 1m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb1f3781486e4d999465eb6fa0dae6cc : SUCCESS in 1h 45m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95eacc786efa4bb7bb63d5f123466888 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd5b52934d8f4d37b3272dd02bb7def1 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b06c58ac0b064515a94527b8335db873 : SUCCESS in 1m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8131c3186a364e8387b9da1c6a0932f9 : SUCCESS in 4m 13s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/a41cadc1f14444e697e4d3a64df83520 : SUCCESS in 17m 18s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/c2e096493f8a494daa682ff91707378d : SUCCESS in 2m 47s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/03dedd5e10a7447fa97b2b041998faf7 : TIMED_OUT in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b42b8e2063695e9d0ca1fa72c981a0d2b1cbb773","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-04 08:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49c82b8d0ab7201cef91bc781556e2436c862871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 10:24:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb52a8515e5341f29bb679c8b3fc5982\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/972942e9aa92432c80d2a8cc3152077c : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb6f13f0e6794592aa047ccbfd11d7be : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfdcfcf3f1904d71a9237b5c2c942b7b : SUCCESS in 1h 54m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f15f19d85848e4b7c599fc68f347a7 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82b32096d1d34195854054a1594c4167 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d410dff79a10486399255cbce6968c27 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9366a7c641a4a94a0784baed89b4e5b : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e357c2f4914a5797813243394a2bab : SUCCESS in 2m 40s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/85a6803ec5004f25936bb4ea1daf8b7d : SUCCESS in 35m 09s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c5c79eec5bc0448a8e1d8c645084939f : SUCCESS in 35m 41s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/287b6cac198646bb8fb8eb88a96a1f5c : SUCCESS in 2m 08s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/d6041e55e8114a05b27300e2cee243e3 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d78825f03f0dc59dbf9060cf69e38598cfd1dc0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 10:24:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e959ba3f15f872390739cf61c438defbd88b7ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:07:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296f99831521412b8663d75f4cfdb5ec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2086b8e0e7934dc2be95f1ef2554b1c0 : SUCCESS in 2m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/650185456da244ae94ba7687f37a4c4f : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4fc529c014e45beab8cdf48181d0808 : SUCCESS in 1h 41m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774b69b70c73424ab10ff50da130430b : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b2753aba35f45d59219eca1f1e508c3 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0c847efae2b4365a7d56eb573ca7dd4 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0d083f7d7d42caab68df09eddf4bd3 : SUCCESS in 3m 22s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/1975cd9c855c40fca1de5ef254cb3c7c : SUCCESS in 35m 26s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/4f094d1ab51a46f6a66870a909e28197 : SUCCESS in 3m 36s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/239223ed16e94933844f77e1efa87e5a : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"db5c682993f27925e3b018f71828a4dd10ffd16b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:07:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62959840232166e4ccf9b696b6b472b9ab2858d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 12:08:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40d0b90c4fc742f6b94f99ea0ab3463f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b4e2934e6c4389af3bd911f1e7b2fa : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f89f0a6443f4fcb27d48163364e1b337e28f4af","revisions":{"4f89f0a6443f4fcb27d48163364e1b337e28f4af":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:19:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/62/979562/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/62/979562/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/62/979562/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/62/979562/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/62/979562/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/62/979562/1"}}},"commit":{"parents":[{"commit":"b2818f131f4365aee3daa092bbae22230c8974b7","subject":"ruff: Configure hacking as external linter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/b2818f131f4365aee3daa092bbae22230c8974b7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 12:18:56.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-09 12:18:56.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: I88beb168eb5a01d5b696cd0da6ca696cdc3149ea\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4f89f0a6443f4fcb27d48163364e1b337e28f4af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/4f89f0a6443f4fcb27d48163364e1b337e28f4af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
