)]}'
{"id":"openstack%2Foslo.log~989761","triplet_id":"openstack%2Foslo.log~master~Ic5660ac3b2349c43ff8a03954d3c3a4c3590187a","project":"openstack/oslo.log","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-28 08:47:41.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-05-28 07:24:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"Ic5660ac3b2349c43ff8a03954d3c3a4c3590187a","subject":"Don\u0027t import eventletutils","status":"MERGED","created":"2026-05-22 12:32:48.000000000","updated":"2026-05-28 08:48:45.000000000","submitted":"2026-05-28 08:47:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989761-eventlet-removal","meta_rev_id":"954b229ad78489a32230fcb500b9ae9cd3108fdf","_number":989761,"virtual_id_number":989761,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 08:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-26 11:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2026-05-26 15:01:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-05-28 07:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2026-05-28 07:24:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 13:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 08:28:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2026-05-26 11:42:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-05-28 07:24:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"85c6978453b794acc48c9586de1a6a5ea1298990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-22 12:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bef0f8342bf3ce9701483109973811b0ed860096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 13:40:43.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f43abc1d22943318057ef84a81f4d0f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7269de491f344a7f94a56fcc459bd9fc : SUCCESS in 1h 06m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109562518cbe40e291ca2f7a74528115 : FAILURE in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/720b63b970d24426b8a0e6965d49ffc2 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f50597c3e93c49eaa999aef1015bbdae : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0ebc9b32428496a8ac992c62e1a5185 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbd0747435c4c50a24ce1c3608cb4e5 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8b6fb22280450dac26947729c7bde0 : SUCCESS in 4m 28s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/7e89e58e5ec44e81a423886b23555ed7 : SUCCESS in 54m 13s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/5a3049bdd69342c08d195abcb3b3d751 : SUCCESS in 38m 55s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/e178b23fa41f481286797f9de2833261 : SUCCESS in 2m 13s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/1427908674704c13b93071c58077e5ae : SUCCESS in 3m 21s\n\nWarning:\n  Comments left for invalid file oslo_log/_options.py\n  Comments left for invalid file oslo_log/formatters.py\n  Comments left for invalid file oslo_log/tests/unit/test_log.py\n  Comments left for invalid file oslo_log/versionutils.py","accounts_in_message":[],"_revision_number":1},{"id":"60d3fb061611cd2b78ee148ab9585be8d332a95a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-26 08:28:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f72ea65a8469f1f4b86a32f33ffd58be178ba67","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-26 08:29:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"866600f4300a2c38c8d99b9e8ec43747e2916d06","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 08:52:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b66e4552d6827a6258d22f82ffc8661b44b7e64a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 08:52:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9d488446b5e46a6cf2dcdb225b027f480f4d01ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 10:09:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e0165787b32433986ef69719eac27f0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baccd6e46c2147d3a7457764c37318af : SUCCESS in 1h 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1749d321a574b2282da320bde82e605 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2215ddbe27704447945f59f0982884ed : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f7f73c9743f4a47ad2052b6490d94f0 : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26829dea4963463290892c72130ec158 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273b01c6a4744696b8ffffbe202e3aa9 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b22eab139cdc4d81867865baac7b5db2 : SUCCESS in 2m 43s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/179a1e4f1ae34626a3316c1bdc18223f : SUCCESS in 46m 11s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/c8a5bc185a8c40a7aaedc4f59c84f18e : SUCCESS in 44m 46s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/cba866616e734e7db223f486956c2ca3 : SUCCESS in 3m 13s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/aded4c3ae7c34c5abbd8562dca98675c : TIMED_OUT in 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"64d28f78021367360c16432e5e0d08e05a42c5b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-26 11:42:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"defde7b76a7f71510971ab7628d26ab49d888f2e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 13:07:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2be0adfb90845d4abcd16ede154c0cfa87f09f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:11:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823cea226bdb4feaa6bdc310bb428cb0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88d53775774d4ae2b564391ee413cde1 : SUCCESS in 1h 27m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571aa990df2c4ac485ceaca89a793ba7 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b87cf387d574e2181b260a8f9e57f59 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29ce0f7e47d3417fbcb9a27389b4a1e2 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54750ed8cb09487dbe24071bf7e7efe1 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60dffb2ab8d543919961df5bc8dc023b : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/313aebaa3171424d803250d4b837c6c2 : SUCCESS in 3m 36s\n- oslo.log-src-grenade https://zuul.opendev.org/t/openstack/build/d54effe4b9744a33bf6fd6e7e8080d18 : SUCCESS in 32m 28s (non-voting)\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/f0312b8f76a04158a3777960ecfd9a61 : SUCCESS in 19m 38s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/40685badb94f468ab12d3c92e7110b7a : SUCCESS in 3m 24s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/d028f16b8d894910a8a8d6640357f260 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8ba8a1b53d4b1a4819d2a9a2dfb4461f983db4d5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 15:01:20.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27d294dfe5b1ad5606f51b271fdceb9c14305ed1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-28 07:24:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c920be0b6cdaaceddd1f28ce8a1a1b5d49f9e69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 07:24:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5828291573afedd0ff5fd53c3263d334eb6ad7f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:47:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9918d927216843dfa22e42ccfa020a13\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/250b3f1e8be84e1a95d77d62d2032073 : SUCCESS in 1h 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fa9594b46446f29b2cc82cb765e62d : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32877857071c4c15ac3c51d27a7ac4b2 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/607fa9fc91e849949dfd9e673419d3b5 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d6fc27c50214f3d964115a1aa04ef26 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f613f33ec9042e2a5d1695fad644740 : SUCCESS in 2m 59s\n- oslo.log-jsonformatter https://zuul.opendev.org/t/openstack/build/56584109b72f4e418d2ae88a68970282 : SUCCESS in 35m 04s\n- oslo.log-tox-py312-eventlet-patched https://zuul.opendev.org/t/openstack/build/278a5335c602417790504aa3108595eb : SUCCESS in 3m 14s\n- oslo.log-tox-py312-eventlet-patched-asyncio https://zuul.opendev.org/t/openstack/build/0f73661c61bd4e49b9782f7761672f59 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dbb4ebb2df577f922b65a8b8edc6aa0bad68497c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:47:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"954b229ad78489a32230fcb500b9ae9cd3108fdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 08:48:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a50725a808430091f362492e7ff8c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0e704afa9649129d632ad0af4fb17a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d78a75c6e7c32c31ab0892c9e3dc7f4cdd724e4b","revisions":{"fc305bb7aa1c5745392f513ec28a8f8811f3d5c2":{"kind":"REWORK","_number":1,"created":"2026-05-22 12:32:48.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/61/989761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/989761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/1"}}},"commit":{"parents":[{"commit":"30c25ef4ae9e6d5f33f6a0f562e47d397e8d01b1","subject":"Merge \"Deprecate version module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/30c25ef4ae9e6d5f33f6a0f562e47d397e8d01b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-22 12:05:45.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-22 12:31:43.000000000","tz":120},"subject":"Don\u0027t import eventletutils","message":"Don\u0027t import eventletutils\n\neventletutils is deprecated and gives extra noise in the\nlogs for every service start.\n\nChange-Id: Ic5660ac3b2349c43ff8a03954d3c3a4c3590187a\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fc305bb7aa1c5745392f513ec28a8f8811f3d5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/fc305bb7aa1c5745392f513ec28a8f8811f3d5c2"}]},"branch":"refs/heads/master"},"d78a75c6e7c32c31ab0892c9e3dc7f4cdd724e4b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-26 08:52:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/61/989761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.log","ref":"refs/changes/61/989761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.log refs/changes/61/989761/2"}}},"commit":{"parents":[{"commit":"0640ed88feed42e01412dcd5d409e024e11175ed","subject":"Fix mypy type checking errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/0640ed88feed42e01412dcd5d409e024e11175ed"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-22 12:05:45.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-05-26 08:52:12.000000000","tz":0},"subject":"Don\u0027t import eventletutils","message":"Don\u0027t import eventletutils\n\neventletutils is deprecated and gives extra noise in the\nlogs for every service start.\n\nChange-Id: Ic5660ac3b2349c43ff8a03954d3c3a4c3590187a\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d78a75c6e7c32c31ab0892c9e3dc7f4cdd724e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.log/commit/d78a75c6e7c32c31ab0892c9e3dc7f4cdd724e4b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"fc305bb7aa1c5745392f513ec28a8f8811f3d5c2","theirs":"0640ed88feed42e01412dcd5d409e024e11175ed","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
