)]}'
{"id":"openstack%2Foslo.messaging~1006763","triplet_id":"openstack%2Foslo.messaging~stable%2F2026.1~Ib04346eef7e88d01e6c6bda79e10708cac553321","project":"openstack/oslo.messaging","branch":"stable/2026.1","topic":"bug/2077152-stable/2026.1","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-30 02:50:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib04346eef7e88d01e6c6bda79e10708cac553321","subject":"Fix rabbit consumer reconnection loop","status":"MERGED","created":"2026-09-22 16:20:21.000000000","updated":"2026-09-30 02:53:26.000000000","submitted":"2026-09-30 02:50:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006763-bug/2077152-stable/2026.1","cherry_pick_of_change":1006175,"cherry_pick_of_patch_set":1,"meta_rev_id":"2a0061e1babfdde24db5f4ee89019f52f6007b64","_number":1006763,"virtual_id_number":1006763,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-30 02:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-09-28 16:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2026-09-30 00:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-22 20:04:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-09-30 00:52:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 16:20:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"state":"CC"},{"updated":"2026-09-22 18:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 20:04:09.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-09-28 16:36:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-09-30 00:52:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"a2805d8b66974f53225aa0f3e239ed9802b09bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-22 16:20:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.2.","accounts_in_message":[],"_revision_number":1},{"id":"6604f28b755290cc4f7bd0f623aafe45b85880a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:19:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b4d6984a774bdebb72a24197f2ef43\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3364a076e1ff4199a911cc569639defd : SUCCESS in 1h 22m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c490941b197144ff853a7b2bd0501bed : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1001b21019c745d082aa1dcef11df24f : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7598e038abb04e62b85510b79872dc6c : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95ea7b1604e443cb9058496304877774 : SUCCESS in 6m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44fbc97e4ff14db3bbac76b1f55cf13a : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a003db748449444280d11383752fe026 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e28d4389a27648358bfad78b4db3eb2c : SUCCESS in 4m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/07413dd930264f619806df77da075afc : SUCCESS in 6m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0860a93981d841cdb7c07854fadfcf41 : FAILURE in 10m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/73a0407521e14ea4a938ab442dfd380c : SUCCESS in 1h 53m 41s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2e72ccda5f6b41d6a4557388e81d9fa0 : SUCCESS in 1h 43m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ed6f63c8aa2b4c85ac192d4a72c5408c : SUCCESS in 58m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/75afa11ebc944bfea0fceda1988753ed : SUCCESS in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1ed80b55c5e3d51567baa5ca0d0e3f08e82ff3c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-22 20:04:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"876caf4fce3aa0f79f2fcfe91bcd396437dd7788","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-28 16:36:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"362921b30f098b297c4004ac5a5d81341c600d56","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-30 00:52:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41e25a644221c660672bc88b54544c1061466b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 00:53:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93d0269d2a97cd928f2ac560fd84fca28b85f172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 02:50:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5618660964045c2b0f0b38857dcfd62\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db4f99437a0e488ca4e93cfd00f0236e : SUCCESS in 1h 53m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60e96b378aa4273b611447aa427441e : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/004d0437c9394cbebcc05d2bb1555ec7 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19712dbcf6e14a50b70b641108d848e4 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b744c0fb04da425080e55ad483ce153c : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/906f2fa998df4540bd5776410b0e64c3 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a52b0d12dd56495883f26969c0403dda : SUCCESS in 5m 49s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/29b1088a59e144cc830608a7465c4ac5 : SUCCESS in 1h 51m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a7a72c41e89e0099f5e0038aded9d49b4ac58e43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 02:50:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a0061e1babfdde24db5f4ee89019f52f6007b64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 02:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2430297f48444c84df371bcaffd1f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a46ba85100149a888c85291e55926cf : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"767919a356c34336ec5a69cac1686d8a87506456","revisions":{"767919a356c34336ec5a69cac1686d8a87506456":{"kind":"REWORK","_number":1,"created":"2026-09-22 16:20:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/1006763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/1006763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/1006763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/1006763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/1006763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/1006763/1"}}},"commit":{"parents":[{"commit":"399f96e8044419ea16929a39174617ba59644052","subject":"Fix RabbitMQ TLS hostname verification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/399f96e8044419ea16929a39174617ba59644052"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-09-22 16:20:21.000000000","tz":0},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy` and\ncalling self.connection.maybe_switch_next().\n\nAssisted-By: claude-opus-4.6\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Nikita Kushnarenko \u003cNKushnarenko@itkey.com\u003e\n(cherry picked from commit b53730d307783ec435a49ce2cb753a6b4f384005)\n(cherry picked from commit 06c0416e5eb6dd82c55fe4310b11c08515fdcac8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/767919a356c34336ec5a69cac1686d8a87506456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/767919a356c34336ec5a69cac1686d8a87506456"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"c43041b8883f92bedd8fe3fdafeb706dc48494a3","ours":"399f96e8044419ea16929a39174617ba59644052","theirs":"06c0416e5eb6dd82c55fe4310b11c08515fdcac8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
