)]}'
{"id":"openstack%2Foslo.messaging~288719","triplet_id":"openstack%2Foslo.messaging~master~I4e29025f33d740bf4bf49a0d1b6ec0495d6a2ea5","project":"openstack/oslo.messaging","branch":"master","topic":"288719","hashtags":[],"change_id":"I4e29025f33d740bf4bf49a0d1b6ec0495d6a2ea5","subject":"Add ability to iterate over dispatchable methods/endpoints","status":"ABANDONED","created":"2016-03-04 20:21:49.000000000","updated":"2018-04-11 22:53:32.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cbfa0bdee8f9c4b8768b53a0328641d6fe5382f9","_number":288719,"virtual_id_number":288719,"owner":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"actions":{},"labels":{"Verified":{"all":[{"date":"2016-03-18 15:37:09.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"date":"2016-04-01 19:55:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-18 15:37:09.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2016-04-01 19:55:04.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"}],"messages":[{"id":"17b65bbf377e7db7e1394f88b537a265e24b7bba","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-03-04 20:21:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"595fbfa708cc1c8f68b4734e96302809ff8497ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-04 22:05:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-pep8/196b155/ : SUCCESS in 56s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/19/288719/1/check/gate-oslo.messaging-docs/2c99498//doc/build/html/ : SUCCESS in 3m 56s\n- gate-oslo.messaging-python27 http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-python27/c326069/ : FAILURE in 1m 43s\n- gate-tempest-dsvm-neutron-src-oslo.messaging http://logs.openstack.org/19/288719/1/check/gate-tempest-dsvm-neutron-src-oslo.messaging/a9615e0/ : SUCCESS in 1h 06m 47s\n- gate-tempest-dsvm-neutron-src-oslo.messaging-liberty http://logs.openstack.org/19/288719/1/check/gate-tempest-dsvm-neutron-src-oslo.messaging-liberty/ea47b9e/ : FAILURE in 9m 59s (non-voting)\n- gate-oslo.messaging-dsvm-functional-rabbit http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-dsvm-functional-rabbit/fa32c54/ : SUCCESS in 17m 33s\n- gate-oslo.messaging-dsvm-functional-amqp1-centos7 http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-dsvm-functional-amqp1-centos7/7915cf6/ : SUCCESS in 34m 33s\n- gate-oslo.messaging-dsvm-functional-pika http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-dsvm-functional-pika/28790d0/ : SUCCESS in 18m 45s (non-voting)\n- gate-oslo.messaging-dsvm-functional-zeromq http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-dsvm-functional-zeromq/78102e1/ : FAILURE in 1h 26m 57s\n- gate-oslo.messaging-python34 http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-python34/a35e3d5/ : FAILURE in 6m 54s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-amqp1-centos7 http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-src-dsvm-full-amqp1-centos7/d422be0/ : SUCCESS in 1h 05m 08s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-pika http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-src-dsvm-full-pika/28dd09b/ : SUCCESS in 48m 06s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-zmq http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-src-dsvm-full-zmq/2a274d8/ : SUCCESS in 40m 13s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-src-grenade-dsvm/1ecfe8c/ : FAILURE in 12m 50s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-src-grenade-dsvm-multinode/b16a6f3/ : FAILURE in 11m 41s (non-voting)\n- gate-telemetry-dsvm-integration-oslo.messaging http://logs.openstack.org/19/288719/1/check/gate-telemetry-dsvm-integration-oslo.messaging/8351149/ : SUCCESS in 32m 32s\n- gate-oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-telemetry-dsvm-integration-pika/0a45d83/ : SUCCESS in 37m 03s (non-voting)\n- gate-oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/19/288719/1/check/gate-oslo.messaging-telemetry-dsvm-integration-zmq/3d5a134/ : FAILURE in 53m 06s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-pika http://logs.openstack.org/19/288719/1/check/gate-tempest-neutron-dsvm-src-oslo.messaging-pika/700a7af/ : SUCCESS in 1h 00m 38s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-zmq http://logs.openstack.org/19/288719/1/check/gate-tempest-neutron-dsvm-src-oslo.messaging-zmq/5271985/ : SUCCESS in 1h 08m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2195a97b4690b505e03083e7a69f367c5d9f5e1c","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-03-07 23:54:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dab3a8617a19e4b69b0906a8d944716078616381","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-08 00:59:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-pep8/7260a49/ : SUCCESS in 2m 59s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/19/288719/2/check/gate-oslo.messaging-docs/7a72f2f//doc/build/html/ : SUCCESS in 1m 34s\n- gate-oslo.messaging-python27 http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-python27/6f97b89/ : FAILURE in 2m 18s\n- gate-oslo.messaging-requirements http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-requirements/d059254/ : SUCCESS in 2m 46s\n- gate-tempest-dsvm-neutron-src-oslo.messaging http://logs.openstack.org/19/288719/2/check/gate-tempest-dsvm-neutron-src-oslo.messaging/44f208a/ : SUCCESS in 59m 06s\n- gate-tempest-dsvm-neutron-src-oslo.messaging-liberty http://logs.openstack.org/19/288719/2/check/gate-tempest-dsvm-neutron-src-oslo.messaging-liberty/087ccd5/ : FAILURE in 10m 06s (non-voting)\n- gate-oslo.messaging-dsvm-functional-rabbit http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-dsvm-functional-rabbit/7042f02/ : SUCCESS in 18m 22s\n- gate-oslo.messaging-dsvm-functional-amqp1-centos7 http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-dsvm-functional-amqp1-centos7/60c4b59/ : NOT_REGISTERED\n- gate-oslo.messaging-dsvm-functional-pika http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-dsvm-functional-pika/a8ed70a/ : SUCCESS in 18m 23s (non-voting)\n- gate-oslo.messaging-dsvm-functional-zeromq http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-dsvm-functional-zeromq/8e90254/ : SUCCESS in 18m 36s\n- gate-oslo.messaging-python34 http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-python34/5f7e660/ : FAILURE in 4m 21s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-amqp1-centos7 http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-src-dsvm-full-amqp1-centos7/626ef92/ : NOT_REGISTERED (non-voting)\n- gate-oslo.messaging-src-dsvm-full-pika http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-src-dsvm-full-pika/b8528af/ : SUCCESS in 44m 46s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-zmq http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-src-dsvm-full-zmq/98e4b05/ : SUCCESS in 40m 57s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-src-grenade-dsvm/3dae76a/ : FAILURE in 16m 44s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-src-grenade-dsvm-multinode/cd311c7/ : NOT_REGISTERED (non-voting)\n- gate-telemetry-dsvm-integration-oslo.messaging http://logs.openstack.org/19/288719/2/check/gate-telemetry-dsvm-integration-oslo.messaging/03e6052/ : SUCCESS in 31m 57s\n- gate-oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-telemetry-dsvm-integration-pika/ce4b470/ : SUCCESS in 23m 05s (non-voting)\n- gate-oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/19/288719/2/check/gate-oslo.messaging-telemetry-dsvm-integration-zmq/4b69e02/ : FAILURE in 32m 33s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-pika http://logs.openstack.org/19/288719/2/check/gate-tempest-neutron-dsvm-src-oslo.messaging-pika/6555ddb/ : SUCCESS in 1h 04m 07s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-zmq http://logs.openstack.org/19/288719/2/check/gate-tempest-neutron-dsvm-src-oslo.messaging-zmq/2b428c8/ : SUCCESS in 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4957064e72e173dbda5656c08ef2d3eb3fc7c513","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-03-08 19:00:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f723d383ef42e209532cd364d8ea94a9dbbcaef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-08 20:06:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-pep8/6346fc7/ : SUCCESS in 1m 47s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/19/288719/3/check/gate-oslo.messaging-docs/25531b5//doc/build/html/ : SUCCESS in 6m 56s\n- gate-oslo.messaging-python27 http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-python27/1c9d2d9/ : SUCCESS in 2m 52s\n- gate-oslo.messaging-requirements http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-requirements/c21f3cf/ : SUCCESS in 2m 51s\n- gate-tempest-dsvm-neutron-src-oslo.messaging http://logs.openstack.org/19/288719/3/check/gate-tempest-dsvm-neutron-src-oslo.messaging/e9b2bd9/ : SUCCESS in 1h 00m 20s\n- gate-tempest-dsvm-neutron-src-oslo.messaging-liberty http://logs.openstack.org/19/288719/3/check/gate-tempest-dsvm-neutron-src-oslo.messaging-liberty/d02b736/ : FAILURE in 7m 02s (non-voting)\n- gate-oslo.messaging-dsvm-functional-rabbit http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-dsvm-functional-rabbit/765709f/ : SUCCESS in 21m 12s\n- gate-oslo.messaging-dsvm-functional-amqp1-centos7 http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-dsvm-functional-amqp1-centos7/d1bbefd/ : SUCCESS in 27m 56s\n- gate-oslo.messaging-dsvm-functional-pika http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-dsvm-functional-pika/10ea7b6/ : SUCCESS in 23m 47s (non-voting)\n- gate-oslo.messaging-dsvm-functional-zeromq http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-dsvm-functional-zeromq/44b6468/ : SUCCESS in 19m 21s\n- gate-oslo.messaging-python34 http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-python34/7058f87/ : FAILURE in 5m 52s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-amqp1-centos7 http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-src-dsvm-full-amqp1-centos7/603c73a/ : SUCCESS in 1h 04m 58s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-pika http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-src-dsvm-full-pika/31fd145/ : SUCCESS in 45m 44s (non-voting)\n- gate-oslo.messaging-src-dsvm-full-zmq http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-src-dsvm-full-zmq/8f1f861/ : SUCCESS in 39m 09s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-src-grenade-dsvm/7b84226/ : FAILURE in 19m 55s (non-voting)\n- gate-oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-src-grenade-dsvm-multinode/c3c4976/ : FAILURE in 15m 45s (non-voting)\n- gate-telemetry-dsvm-integration-oslo.messaging http://logs.openstack.org/19/288719/3/check/gate-telemetry-dsvm-integration-oslo.messaging/021005d/ : SUCCESS in 23m 41s\n- gate-oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-telemetry-dsvm-integration-pika/c484d57/ : FAILURE in 42m 54s (non-voting)\n- gate-oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/19/288719/3/check/gate-oslo.messaging-telemetry-dsvm-integration-zmq/934a740/ : FAILURE in 45m 56s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-pika http://logs.openstack.org/19/288719/3/check/gate-tempest-neutron-dsvm-src-oslo.messaging-pika/65cfadb/ : SUCCESS in 58m 01s (non-voting)\n- gate-tempest-neutron-dsvm-src-oslo.messaging-zmq http://logs.openstack.org/19/288719/3/check/gate-tempest-neutron-dsvm-src-oslo.messaging-zmq/b7b7196/ : SUCCESS in 58m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3152ff6c874f96ae572b9637db1403c88388a5d9","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-03-10 11:43:42.000000000","message":"Patch Set 3: Code-Review-1\n\nsome tests please? possibly functional tests?","accounts_in_message":[],"_revision_number":3},{"id":"63c6ea0567a90bdda415bcd238e3baf13dcd6ded","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2016-03-18 15:37:09.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n+1 on tests.","accounts_in_message":[],"_revision_number":3},{"id":"42db2c40725983b10d26d66facbbd9173d5f752e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-30 15:36:27.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"07240a2c7b29d8f5608dd0b5399fa3addcff5b66","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2016-04-01 19:55:04.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"0e3ad6e28d266dd758b2b5f04cde3bb0ebea874d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 05:20:35.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":3},{"id":"2cf5bb47e78c7d219b0dc87d2c890c56f3de742a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2018-04-11 22:53:32.000000000","message":"Abandoned\n\nThis has been in merge conflict for two years.  Please feel free to restore if you still want it.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"537ef2197ba283f023241e107ba07bfe989bd114","revisions":{"14734bf062f130f1ae610cc04071a24e6a8d6680":{"kind":"REWORK","_number":1,"created":"2016-03-04 20:21:49.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/19/288719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/288719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/1"}}},"commit":{"parents":[{"commit":"5b75f3967681e2c88e94ca94d54856a27f4713b5","subject":"Merge \"Simulator: fix batch-notify-server command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5b75f3967681e2c88e94ca94d54856a27f4713b5"}]}],"author":{"name":"Joshua Harlow","email":"harlowja@gmail.com","date":"2016-03-04 20:16:28.000000000","tz":-480},"committer":{"name":"Joshua Harlow","email":"harlowja@gmail.com","date":"2016-03-04 20:21:38.000000000","tz":-480},"subject":"Add rpc \u0027__stats__\u0027 response","message":"Add rpc \u0027__stats__\u0027 response\n\nAdd a special message method that can be responded to with\ndetails about the dispatcher, including what endpoints it\ncan call, what targets each endpoint has and details about\nthe endpoint that can be useful for introspecting a target\nfrom a remote system.\n\nChange-Id: I4e29025f33d740bf4bf49a0d1b6ec0495d6a2ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14734bf062f130f1ae610cc04071a24e6a8d6680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14734bf062f130f1ae610cc04071a24e6a8d6680"}]},"branch":"refs/heads/master"},"6e8a74de18714e187774b1f5edd505622b2d2115":{"kind":"REWORK","_number":2,"created":"2016-03-07 23:54:38.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/19/288719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/288719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/2"}}},"commit":{"parents":[{"commit":"5b75f3967681e2c88e94ca94d54856a27f4713b5","subject":"Merge \"Simulator: fix batch-notify-server command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5b75f3967681e2c88e94ca94d54856a27f4713b5"}]}],"author":{"name":"Joshua Harlow","email":"harlowja@gmail.com","date":"2016-03-04 20:16:28.000000000","tz":-480},"committer":{"name":"Joshua Harlow","email":"harlowja@yahoo-inc.com","date":"2016-03-07 23:54:37.000000000","tz":-480},"subject":"Add ability to iterate over dispatchable methods/endpoints","message":"Add ability to iterate over dispatchable methods/endpoints\n\nAdd a iter_endpoints() method that can be responded to with\ndetails about the dispatcher, including what endpoints it\ncan call, what targets each endpoint has and details about\nthe endpoint that can be useful for introspecting a server.\n\nTo get this info from a server all that should be needed\nis to do:\n\n  server \u003d get_rpc_server(...)\n  for endpoint, target, methods in server.dispatcher.iter_endpoints():\n     ....\n\nChange-Id: I4e29025f33d740bf4bf49a0d1b6ec0495d6a2ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e8a74de18714e187774b1f5edd505622b2d2115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e8a74de18714e187774b1f5edd505622b2d2115"}]},"branch":"refs/heads/master"},"537ef2197ba283f023241e107ba07bfe989bd114":{"kind":"REWORK","_number":3,"created":"2016-03-08 19:00:21.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/19/288719/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/288719/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/288719/3"}}},"commit":{"parents":[{"commit":"5b75f3967681e2c88e94ca94d54856a27f4713b5","subject":"Merge \"Simulator: fix batch-notify-server command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5b75f3967681e2c88e94ca94d54856a27f4713b5"}]}],"author":{"name":"Joshua Harlow","email":"harlowja@gmail.com","date":"2016-03-04 20:16:28.000000000","tz":-480},"committer":{"name":"Joshua Harlow","email":"harlowja@yahoo-inc.com","date":"2016-03-08 19:00:02.000000000","tz":-480},"subject":"Add ability to iterate over dispatchable methods/endpoints","message":"Add ability to iterate over dispatchable methods/endpoints\n\nAdd a iter_endpoints() method that can be responded to with\ndetails about the dispatcher, including what endpoints it\ncan call, what targets each endpoint has and details about\nthe endpoint that can be useful for introspecting a server.\n\nTo get this info from a server all that should be needed\nis to do:\n\n  server \u003d get_rpc_server(...)\n  for endpoint, target, methods in server.dispatcher.iter_endpoints():\n     ....\n\nChange-Id: I4e29025f33d740bf4bf49a0d1b6ec0495d6a2ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/537ef2197ba283f023241e107ba07bfe989bd114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/537ef2197ba283f023241e107ba07bfe989bd114"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
