)]}'
{"id":"openstack%2Foslo.messaging~546763","triplet_id":"openstack%2Foslo.messaging~master~I60334aaf019f177a984583528b71d00859d31f84","project":"openstack/oslo.messaging","branch":"master","topic":"heartbeat-active-calls","hashtags":[],"change_id":"I60334aaf019f177a984583528b71d00859d31f84","subject":"[rabbitmq] Implement active call monitoring","status":"MERGED","created":"2018-02-21 20:17:31.000000000","updated":"2018-05-16 19:25:21.000000000","submitted":"2018-05-16 19:25:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"546763-1526498721433-49d6adcc","meta_rev_id":"d5f2b988281c956a39a61eff744b112c3b92e977","_number":546763,"virtual_id_number":546763,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2018-05-09 17:09:45.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2018-05-16 19:25:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2018-05-10 13:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2018-05-16 17:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2018-05-16 17:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-27 09:39:44.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2018-02-27 09:39:55.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2018-05-10 13:42:52.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2018-05-16 17:51:06.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2018-05-16 19:25:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4c778d2af69c7751eaf7ccfe7d00ecc7d209ebe","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-02-21 20:17:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b6c9eaedb2e4a6df45f7be730db081a724cc3df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 21:57:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/1/check/openstack-tox-pep8/44bea06/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/1/check/openstack-tox-py27/86a4144/ : FAILURE in 14m 35s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/1/check/openstack-tox-py35/e71f731/ : TIMED_OUT in 40m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/1/check/build-openstack-sphinx-docs/34ed35d/html/ : SUCCESS in 3m 05s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/1/check/legacy-tempest-dsvm-neutron-src/73345e6/ : SUCCESS in 1h 26m 39s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-amqp1/e11f652/ : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-kafka/66674e9/ : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-pika/79e99f0/ : FAILURE in 5m 34s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-rabbit/de6166d/ : FAILURE in 5m 02s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-zmq-proxy/6557be2/ : TIMED_OUT in 40m 39s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-zmq-pubsub/9b186a7/ : TIMED_OUT in 40m 38s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py27-func-zmq/18f2010/ : TIMED_OUT in 40m 43s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py35-func-amqp1/11c2ff1/ : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py35-func-rabbit/14c26b9/ : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/63/546763/1/check/oslo.messaging-tox-py35-func-zmq/2a058e4/ : TIMED_OUT in 40m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-rabbit-default/6c333d6/ : SUCCESS in 1h 33m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/0404ee3/ : FAILURE in 22m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/9383713/ : FAILURE in 30m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/5e6d25e/ : FAILURE in 36m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-kafka-default/f54b1e7/ : FAILURE in 22m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-pika-default/d9a79e6/ : FAILURE in 32m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-dsvm-full-zmq-default/0efdb25/ : FAILURE in 14m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-grenade-dsvm/87835f4/ : SUCCESS in 1h 01m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/1/check/oslo.messaging-src-grenade-dsvm-multinode/e44294e/ : SUCCESS in 55m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/8dd9e89/ : SUCCESS in 34m 58s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c931ce6/ : FAILURE in 46m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/797f1a1/ : FAILURE in 41m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/63/546763/1/check/oslo.messaging-telemetry-dsvm-integration-pika/6187937/ : FAILURE in 33m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/63/546763/1/check/oslo.messaging-telemetry-dsvm-integration-zmq/9776919/ : FAILURE in 19m 25s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/159fa5c/ : SUCCESS in 1h 15m 53s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/1/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/e6cd813/ : FAILURE in 22m 13s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/78fe635/ : FAILURE in 25m 01s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/99f0491/ : FAILURE in 21m 54s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/63/546763/1/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/e863e51/ : FAILURE in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32c460c744a9d55d736137ac296835ba4f3ea644","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-02-27 09:39:01.000000000","message":"Patch Set 1:\n\n+1 to the feature.\n\nas discussed at PTG, we\u0027ll want the Client constructor API changed to enable this on a per-client basis and a flag on the call method itself to opt in/out on a per call basis.\n\nIf you\u0027re willing to update the APIs to your liking I can help out with the driver changes.","accounts_in_message":[],"_revision_number":1},{"id":"a4aa5afb54b13dc1abffcc1bcc3744c7ffa81804","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-03 22:26:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a0b74a94bf8f42b095597dd98946da48af13f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-03 23:48:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/2/check/openstack-tox-pep8/5562ea9/ : FAILURE in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/2/check/openstack-tox-py27/9338ee4/ : FAILURE in 7m 49s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/2/check/openstack-tox-py35/d4f9bc2/ : FAILURE in 7m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/2/check/build-openstack-sphinx-docs/388092d/html/ : SUCCESS in 3m 36s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/2/check/legacy-tempest-dsvm-neutron-src/42895ed/ : SUCCESS in 1h 12m 37s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-amqp1/a99256d/ : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-kafka/c80f331/ : FAILURE in 4m 59s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-pika/bc690fb/ : FAILURE in 4m 14s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-rabbit/0da821d/ : FAILURE in 4m 26s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-zmq-proxy/cd93e78/ : TIMED_OUT in 40m 46s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-zmq-pubsub/61c307a/ : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py27-func-zmq/c6bfcba/ : TIMED_OUT in 40m 43s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py35-func-amqp1/125d7f3/ : FAILURE in 3m 58s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py35-func-rabbit/d59c550/ : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/63/546763/2/check/oslo.messaging-tox-py35-func-zmq/89cde60/ : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-rabbit-default/c9a1893/ : SUCCESS in 1h 12m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/ca397f7/ : FAILURE in 29m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/58b5589/ : FAILURE in 34m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/8496936/ : FAILURE in 35m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-kafka-default/c0f0390/ : FAILURE in 25m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-pika-default/10a67aa/ : FAILURE in 22m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-dsvm-full-zmq-default/0ecf050/ : FAILURE in 14m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-grenade-dsvm/7e34df6/ : SUCCESS in 45m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/2/check/oslo.messaging-src-grenade-dsvm-multinode/281aee3/ : SUCCESS in 1h 01m 37s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/881358e/ : SUCCESS in 44m 02s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/725b7bd/ : FAILURE in 50m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/5f87e52/ : FAILURE in 36m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/63/546763/2/check/oslo.messaging-telemetry-dsvm-integration-pika/7078bbc/ : FAILURE in 27m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/63/546763/2/check/oslo.messaging-telemetry-dsvm-integration-zmq/34b84f6/ : FAILURE in 15m 20s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/b793e7a/ : SUCCESS in 1h 19m 06s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/2/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/3ff2056/ : FAILURE in 26m 24s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/1518b98/ : FAILURE in 31m 40s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/905ce16/ : FAILURE in 27m 15s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/63/546763/2/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/a1ec655/ : FAILURE in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab78a7cb6a7107bcb0c106db49efdd28b1d18202","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-07 20:17:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2b18d90a10ad3dfd181f9efb245ee6a4d0d92ed4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-07 22:02:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/3/check/openstack-tox-pep8/29ecd8b/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/3/check/openstack-tox-py27/fc0394c/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/3/check/openstack-tox-py35/59e012f/ : SUCCESS in 4m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/3/check/build-openstack-sphinx-docs/edb32b2/html/ : SUCCESS in 2m 52s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/3/check/legacy-tempest-dsvm-neutron-src/e3ee1d4/ : SUCCESS in 1h 29m 15s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-amqp1/5799f9b/ : FAILURE in 4m 21s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-kafka/b887c8b/ : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-pika/4373fbd/ : FAILURE in 4m 16s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-rabbit/3ae0f93/ : SUCCESS in 4m 41s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-zmq-proxy/31d47a8/ : TIMED_OUT in 40m 49s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-zmq-pubsub/516c3d3/ : FAILURE in 5m 14s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py27-func-zmq/0ed4566/ : TIMED_OUT in 40m 45s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py35-func-amqp1/c42b214/ : FAILURE in 3m 39s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py35-func-rabbit/e63e8fc/ : FAILURE in 4m 36s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/63/546763/3/check/oslo.messaging-tox-py35-func-zmq/4cec0fb/ : TIMED_OUT in 40m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-rabbit-default/3aced14/ : SUCCESS in 1h 38m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/25b3d77/ : FAILURE in 29m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/2636609/ : FAILURE in 31m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/eceb826/ : FAILURE in 32m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-kafka-default/3baaf74/ : FAILURE in 26m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-pika-default/3b902dc/ : FAILURE in 22m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-dsvm-full-zmq-default/334c016/ : FAILURE in 15m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-grenade-dsvm/495bfa8/ : SUCCESS in 44m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/3/check/oslo.messaging-src-grenade-dsvm-multinode/b2cb77f/ : SUCCESS in 1h 07m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/1ec8fab/ : SUCCESS in 31m 52s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/ebb57ef/ : FAILURE in 51m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/68213a0/ : FAILURE in 31m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/63/546763/3/check/oslo.messaging-telemetry-dsvm-integration-pika/a5dad8c/ : FAILURE in 37m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/63/546763/3/check/oslo.messaging-telemetry-dsvm-integration-zmq/f4eb24e/ : FAILURE in 13m 05s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/0791fb6/ : SUCCESS in 1h 16m 16s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/3/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/fda7de1/ : FAILURE in 22m 11s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/1044a20/ : FAILURE in 38m 40s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/bb89067/ : FAILURE in 28m 33s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/63/546763/3/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/c7ee163/ : FAILURE in 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"96e4238375d0972244b083585c91f3c701c5fb46","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-03-08 19:51:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"26b8c96b70f4eb89335e527bcf859be8eca99f9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-08 21:37:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/4/check/openstack-tox-pep8/5eebba7/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/4/check/openstack-tox-py27/c96a647/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/4/check/openstack-tox-py35/72da214/ : SUCCESS in 4m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/4/check/build-openstack-sphinx-docs/9a6151e/html/ : SUCCESS in 2m 44s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/4/check/legacy-tempest-dsvm-neutron-src/f1a325d/ : SUCCESS in 1h 14m 00s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-amqp1/eab6e80/ : SUCCESS in 3m 22s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-kafka/c367ef4/ : FAILURE in 4m 23s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-pika/099da1f/ : SUCCESS in 3m 57s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-rabbit/a01078b/ : SUCCESS in 3m 47s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-zmq-proxy/26fb281/ : TIMED_OUT in 40m 45s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-zmq-pubsub/34a8f1c/ : FAILURE in 4m 04s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py27-func-zmq/8dcec6c/ : TIMED_OUT in 40m 41s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py35-func-amqp1/0495614/ : SUCCESS in 3m 55s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py35-func-rabbit/fae729a/ : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/63/546763/4/check/oslo.messaging-tox-py35-func-zmq/14cf257/ : TIMED_OUT in 40m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-rabbit-default/2e7d381/ : SUCCESS in 1h 15m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/03e459c/ : SUCCESS in 1h 32m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/7d1ad2d/ : FAILURE in 28m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/2dfebba/ : FAILURE in 42m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-kafka-default/95e485c/ : SUCCESS in 1h 39m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-pika-default/c4a0182/ : SUCCESS in 1h 29m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-dsvm-full-zmq-default/15fd7ca/ : FAILURE in 12m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-grenade-dsvm/219fce5/ : SUCCESS in 54m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/4/check/oslo.messaging-src-grenade-dsvm-multinode/09d5ca6/ : SUCCESS in 56m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/a59c696/ : SUCCESS in 36m 04s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/77d89dc/ : FAILURE in 47m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/d10fbed/ : FAILURE in 39m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/63/546763/4/check/oslo.messaging-telemetry-dsvm-integration-pika/ba57749/ : FAILURE in 58m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/63/546763/4/check/oslo.messaging-telemetry-dsvm-integration-zmq/0f055fa/ : FAILURE in 13m 32s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/d5c5e72/ : FAILURE in 1h 03m 10s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/4/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/0f65cc7/ : SUCCESS in 1h 23m 27s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/de0fe2d/ : SUCCESS in 1h 11m 54s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/3861539/ : SUCCESS in 1h 30m 12s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/63/546763/4/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/670eb31/ : FAILURE in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af39edefbc6ecb1c005af2a497dc19935cdac69b","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-03-14 20:52:42.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThis will need a new functional test that implements a long running RPC call  (see oslo_messaging/tests/functional).","accounts_in_message":[],"_revision_number":4},{"id":"fcca3b599d31c1f3192ac7e4529d99bf2aa8ce39","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-04-24 14:44:46.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nSorry I\u0027ve been away from this for so long. Hoping to get the remaining bits finished soon.","accounts_in_message":[],"_revision_number":4},{"id":"c9aaaab5d9f78ef8f817123f7bc25f31f251d3ad","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-04-24 15:16:56.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nHmm - I think I misunderstood the proposed API modifications - see inline,\n\nthanks","accounts_in_message":[],"_revision_number":4},{"id":"77b81fe33251d4b85069374f334f97ec3fa5741b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-04-24 16:04:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7e9d03108c1c6613920e53cdf60b1e0ec0c7f89","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-04-24 16:51:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"90ee673e30c6e7a04848bbaaa30cf2cf30183f46","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-04-24 16:51:56.000000000","message":"Patch Set 5: Workflow-1\n\nNot all the tests will run for me locally, so I pushed this up to let zuul have a crack at it.","accounts_in_message":[],"_revision_number":5},{"id":"7aa665f104797fcb14e91c7bed34f20d213cc7ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 20:23:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/5/check/openstack-tox-pep8/6408443/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/5/check/openstack-tox-py27/d426b55/ : TIMED_OUT in 42m 15s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/5/check/openstack-tox-py35/d4c4933/ : TIMED_OUT in 42m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/5/check/build-openstack-sphinx-docs/3f0cbc2/html/ : SUCCESS in 2m 54s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/5/check/legacy-tempest-dsvm-neutron-src/3fa4230/ : FAILURE in 1h 10m 38s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/5/check/oslo.messaging-tox-py27-func-amqp1/80c8a70/ : SUCCESS in 3m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/5/check/oslo.messaging-tox-py27-func-kafka/41ab336/ : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/5/check/oslo.messaging-tox-py27-func-rabbit/57eaac7/ : SUCCESS in 4m 35s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/5/check/oslo.messaging-tox-py35-func-amqp1/be2444e/ : SUCCESS in 3m 50s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/5/check/oslo.messaging-tox-py35-func-rabbit/790a759/ : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-dsvm-full-rabbit-default/bd13097/ : SUCCESS in 1h 31m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/3451b78/ : SUCCESS in 1h 27m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/95c7734/ : SUCCESS in 2h 00m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/b87228b/ : SUCCESS in 1h 26m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-dsvm-full-kafka-default/5d55eb2/ : SUCCESS in 1h 27m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-grenade-dsvm/f6c9225/ : SUCCESS in 1h 01m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/5/check/oslo.messaging-src-grenade-dsvm-multinode/9de2dde/ : SUCCESS in 1h 29m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/5/check/oslo.messaging-telemetry-dsvm-integration-rabbit/a5662af/ : SUCCESS in 32m 45s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/5/check/oslo.messaging-telemetry-dsvm-integration-amqp1/f4f5b4a/ : FAILURE in 1h 02m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/5/check/oslo.messaging-telemetry-dsvm-integration-kafka/ecf1aa5/ : FAILURE in 48m 50s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/5/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/5827141/ : SUCCESS in 1h 28m 14s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/5/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/41e48a6/ : SUCCESS in 1h 14m 45s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/5/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/0f8cdf1/ : SUCCESS in 1h 26m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/5/check/openstack-tox-lower-constraints/f022a23/ : TIMED_OUT in 42m 13s","accounts_in_message":[],"_revision_number":5},{"id":"5bbf8f39473e49e580cf814e45a5213d229cd607","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-02 16:50:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b6bbb133969249e45aafb252db2ae79b00dd3f4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 19:10:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/6/check/openstack-tox-pep8/1bb68df/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/6/check/openstack-tox-py27/ba01342/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/6/check/openstack-tox-py35/815f469/ : SUCCESS in 4m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/6/check/build-openstack-sphinx-docs/6d33c8e/html/ : SUCCESS in 3m 41s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/6/check/legacy-tempest-dsvm-neutron-src/e82b063/ : SUCCESS in 1h 28m 11s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/6/check/oslo.messaging-devstack-amqp1/6dad0ab/ : SUCCESS in 1h 26m 42s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py27-func-amqp1/80fb94a/ : SUCCESS in 4m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py27-func-kafka/0bfadbe/ : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py27-func-rabbit/8aaf73c/ : FAILURE in 5m 17s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py35-func-amqp1/733b81c/ : SUCCESS in 4m 00s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py35-func-kafka/d743c52/ : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/6/check/oslo.messaging-tox-py35-func-rabbit/28c00b3/ : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-dsvm-full-rabbit-default/6f45c8c/ : SUCCESS in 1h 22m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/8515492/ : FAILURE in 1h 08m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/0f7f1bd/ : SUCCESS in 1h 26m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/cf351ec/ : SUCCESS in 1h 31m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-dsvm-full-kafka-default/28d8d3e/ : SUCCESS in 1h 28m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-grenade-dsvm/c7d95da/ : SUCCESS in 52m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/6/check/oslo.messaging-src-grenade-dsvm-multinode/65d55de/ : SUCCESS in 53m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/6/check/oslo.messaging-telemetry-dsvm-integration-rabbit/053faeb/ : SUCCESS in 43m 19s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/6/check/oslo.messaging-telemetry-dsvm-integration-amqp1/d04c3a3/ : FAILURE in 52m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/6/check/oslo.messaging-telemetry-dsvm-integration-kafka/7cf7191/ : FAILURE in 51m 52s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/6/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/e4eaa29/ : SUCCESS in 1h 34m 28s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/6/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/051671f/ : SUCCESS in 1h 22m 33s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/6/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/1821a7a/ : FAILURE in 2h 05m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/6/check/openstack-tox-lower-constraints/dd155a9/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":6},{"id":"67c60ae737c14279da0d458733af0596e1346d38","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-03 14:00:17.000000000","message":"Patch Set 6: Workflow-1\n\nunit tests working now, gotta get back to the func test","accounts_in_message":[],"_revision_number":6},{"id":"6dfb43724ce3ad5f51de4d556acab9fe01e1dae5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-03 16:28:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b9bce1e53a15d2594fdb6e48cdcb35d67334e829","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-03 16:39:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4f7a6034f3094f36f7813b818eca923b477466c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-03 18:21:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/8/check/openstack-tox-pep8/d20eb3f/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/8/check/openstack-tox-py27/be00530/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/8/check/openstack-tox-py35/70e2774/ : SUCCESS in 4m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/8/check/build-openstack-sphinx-docs/3da46cb/html/ : SUCCESS in 3m 10s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/8/check/legacy-tempest-dsvm-neutron-src/595801c/ : SUCCESS in 1h 33m 18s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/8/check/oslo.messaging-devstack-amqp1/5f406db/ : SUCCESS in 1h 30m 45s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py27-func-amqp1/fc1f54f/ : SUCCESS in 5m 07s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py27-func-kafka/450836e/ : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py27-func-rabbit/194f418/ : SUCCESS in 4m 09s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py35-func-amqp1/13e5dae/ : SUCCESS in 3m 21s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py35-func-kafka/68c5342/ : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/8/check/oslo.messaging-tox-py35-func-rabbit/8628872/ : FAILURE in 4m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-dsvm-full-rabbit-default/7e4a8a3/ : SUCCESS in 1h 14m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/2b4679d/ : SUCCESS in 1h 31m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/be93b60/ : SUCCESS in 1h 33m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/514c44b/ : FAILURE in 1h 16m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-dsvm-full-kafka-default/ff74908/ : SUCCESS in 1h 16m 49s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-grenade-dsvm/bcbcca3/ : SUCCESS in 51m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/8/check/oslo.messaging-src-grenade-dsvm-multinode/95fcb11/ : SUCCESS in 55m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/8/check/oslo.messaging-telemetry-dsvm-integration-rabbit/9ea3a6e/ : SUCCESS in 35m 44s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/8/check/oslo.messaging-telemetry-dsvm-integration-amqp1/b4fd546/ : FAILURE in 1h 03m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/8/check/oslo.messaging-telemetry-dsvm-integration-kafka/3154702/ : FAILURE in 48m 52s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/8/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/126d8f8/ : SUCCESS in 1h 20m 02s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/8/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/49f46ee/ : SUCCESS in 1h 38m 59s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/8/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/daa0d15/ : SUCCESS in 1h 21m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/8/check/openstack-tox-lower-constraints/ec25c44/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":8},{"id":"0f9fd705fdf230ae88071d374bee96968e58a00a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-05-04 18:25:06.000000000","message":"Patch Set 8:\n\nDan this looks great.  Once this lands I\u0027ll add support for the remaining drivers.\n\nOne subjective suggestion - after hacking around with this patch I felt that the term \"heartbeat\" has become overloaded in the rabbit code base.  Prior to this patch we had lots of rabbit code and options using the generic term \u0027heartbeat\u0027 for connection monitoring.  Using the term \u0027heartbeat\u0027 to also refer to RPC timeout handling makes the codebase a little ambiguous.  Just IMHO.\n\nI\u0027ve got no problems using the term heartbeat for the client API parameters.  I\u0027m really just inferring to the internal codebase\u0027s use of \u0027heartbeat\u0027.\n\nWhat do you think of a follow on patch with something a little more specific to this feature, like rpc_keepalive or similar for naming methods/attributes?","accounts_in_message":[],"_revision_number":8},{"id":"8fe15e4d420da7e661cda65e27ac1299fc8cecff","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-07 14:10:17.000000000","message":"Patch Set 8:\n\n\u003e Dan this looks great.  Once this lands I\u0027ll add support for the\n \u003e remaining drivers.\n \u003e \n \u003e One subjective suggestion - after hacking around with this patch I\n \u003e felt that the term \"heartbeat\" has become overloaded in the rabbit\n \u003e code base.  Prior to this patch we had lots of rabbit code and\n \u003e options using the generic term \u0027heartbeat\u0027 for connection\n \u003e monitoring.  Using the term \u0027heartbeat\u0027 to also refer to RPC\n \u003e timeout handling makes the codebase a little ambiguous.  Just IMHO.\n \u003e \n \u003e I\u0027ve got no problems using the term heartbeat for the client API\n \u003e parameters.  I\u0027m really just inferring to the internal codebase\u0027s\n \u003e use of \u0027heartbeat\u0027.\n \u003e \n \u003e What do you think of a follow on patch with something a little more\n \u003e specific to this feature, like rpc_keepalive or similar for naming\n \u003e methods/attributes?\n\nThat\u0027s fine with me. I\u0027m happy to rework this patch to use a different name, and I\u0027m also happy to have it exposed via the api with that name. Not sure what that should be though. More word brainstorming:\n\n\"Call signaling\"\n\"Call monitoring\"\n\"Active calls\"\n\"Live call monitoring\"\n\nAnyway, you pick and I\u0027ll rename :)","accounts_in_message":[],"_revision_number":8},{"id":"13c8832b7fdf685ee21c3b3a2cdb76982971d7e9","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-05-08 15:26:19.000000000","message":"Patch Set 8:\n\n\"call monitoring\" +1","accounts_in_message":[],"_revision_number":8},{"id":"7b23bb7432ecf131d6608ea92b742cb4e03ab7a9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-08 15:59:56.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"217c6982d2e9a77efe85e92f774e30e45bb74afc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-08 16:24:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"daa54ef10c1c1563ee6c70462a37cb763dcbf537","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-08 17:09:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"35b068940875b75cb75ca3db86e540c448ae9893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 19:35:10.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/9/check/openstack-tox-pep8/b2eb30e/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/9/check/openstack-tox-py27/9cc43ca/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/9/check/openstack-tox-py35/9557fc5/ : SUCCESS in 4m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/9/check/build-openstack-sphinx-docs/12e0eaa/html/ : SUCCESS in 3m 16s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/9/check/legacy-tempest-dsvm-neutron-src/70cee2b/ : FAILURE in 1h 01m 41s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-devstack-amqp1/44fc9db/ : SUCCESS in 1h 27m 40s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-amqp1/e9c5394/ : SUCCESS in 3m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-kafka/6c1a5aa/ : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-rabbit/c845307/ : SUCCESS in 4m 45s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-amqp1/98d27cb/ : SUCCESS in 4m 45s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-kafka/9f41a17/ : FAILURE in 4m 58s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-rabbit/9d10e04/ : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-rabbit-default/accbc4c/ : SUCCESS in 1h 14m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/eda5f69/ : SUCCESS in 1h 18m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/3c0acae/ : SUCCESS in 1h 27m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/565700d/ : SUCCESS in 1h 32m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default/fa605d4/ : SUCCESS in 1h 20m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm/17a1b36/ : SUCCESS in 50m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm-multinode/1e203e6/ : SUCCESS in 56m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-rabbit/c267092/ : SUCCESS in 37m 14s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-amqp1/0b13ae2/ : FAILURE in 51m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-kafka/ca2b592/ : FAILURE in 53m 33s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/3fb8858/ : SUCCESS in 1h 33m 47s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/044167d/ : FAILURE in 1h 01m 34s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/51a9e1c/ : SUCCESS in 1h 14m 48s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/9/check/openstack-tox-lower-constraints/1cd36bd/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":9},{"id":"948b480f5374169eb4b6cfba5d4f985e161e2ee5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-08 20:56:15.000000000","message":"Patch Set 9:\n\nrecheck doesn\u0027t look like anything related in the logs","accounts_in_message":[],"_revision_number":9},{"id":"b7ea47a241ca24cba9ea73022b20d01ff24a9283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 23:03:43.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/9/check/openstack-tox-pep8/f8b61ba/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/9/check/openstack-tox-py27/af9cfea/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/9/check/openstack-tox-py35/c2c17b3/ : SUCCESS in 4m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/9/check/build-openstack-sphinx-docs/c9021b1/html/ : SUCCESS in 3m 05s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/9/check/legacy-tempest-dsvm-neutron-src/f5515dd/ : SUCCESS in 1h 26m 57s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-devstack-amqp1/118aa42/ : TIMED_OUT in 2h 05m 35s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-amqp1/1249d15/ : SUCCESS in 4m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-kafka/e74c306/ : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-rabbit/01220be/ : SUCCESS in 4m 14s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-amqp1/38f7351/ : SUCCESS in 3m 46s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-kafka/66e600d/ : FAILURE in 6m 02s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-rabbit/397db9c/ : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-rabbit-default/53cc1fb/ : SUCCESS in 1h 36m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/010c746/ : SUCCESS in 1h 16m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/6f4bb77/ : SUCCESS in 1h 54m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/7cf7179/ : SUCCESS in 1h 32m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default/4df5956/ : SUCCESS in 1h 27m 33s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm/bd52545/ : SUCCESS in 45m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm-multinode/9a7da9f/ : SUCCESS in 1h 20m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-rabbit/6a5dab3/ : SUCCESS in 33m 45s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c714797/ : FAILURE in 47m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-kafka/fa107c4/ : FAILURE in 54m 05s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/19899e6/ : SUCCESS in 1h 43m 59s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/480c6a5/ : SUCCESS in 1h 47m 57s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/e2bee40/ : FAILURE in 58m 06s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/9/check/openstack-tox-lower-constraints/a07cd58/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":9},{"id":"c33522d4b352c963c67d2c3f51858a09721f9d1a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-05-09 13:00:01.000000000","message":"Patch Set 9:\n\nUgh - that devstack-amqp1 test is supposed to be non-voting.\n\nCan you change the following line in .zuul.yaml:\n\nhttps://git.openstack.org/cgit/openstack/oslo.messaging/tree/.zuul.yaml#n257\n\nto be:\n\n        - oslo.messaging-devstack-amqp1:\n            voting: false\n\nI\u0027m ok with piggybacking that change into this patch","accounts_in_message":[],"_revision_number":9},{"id":"62bb3678488f6fbe649e6b953ccd40d71635436f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-09 14:29:36.000000000","message":"Patch Set 9:\n\nrecheck kinda looks like tempest lost ability to hit the apis, timed out with all http stuff long after the last things in the server logs?","accounts_in_message":[],"_revision_number":9},{"id":"2aa6f913b93ef7810fbfb09570910d12525b10e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-09 16:27:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/9/check/openstack-tox-pep8/ac46cc2/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/9/check/openstack-tox-py27/1eb2409/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/9/check/openstack-tox-py35/23c62d3/ : SUCCESS in 4m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/9/check/build-openstack-sphinx-docs/bd56630/html/ : SUCCESS in 3m 19s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/9/check/legacy-tempest-dsvm-neutron-src/f68c57f/ : SUCCESS in 1h 21m 35s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-devstack-amqp1/22c70e0/ : SUCCESS in 1h 31m 12s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-amqp1/d306152/ : SUCCESS in 3m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-kafka/625e259/ : FAILURE in 4m 29s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py27-func-rabbit/4d8b764/ : SUCCESS in 4m 00s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-amqp1/6e62f13/ : SUCCESS in 4m 11s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-kafka/206e95f/ : FAILURE in 5m 01s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-tox-py35-func-rabbit/02b6289/ : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-rabbit-default/9dd99af/ : SUCCESS in 1h 20m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/defaa26/ : SUCCESS in 1h 26m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/88400ad/ : SUCCESS in 1h 34m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/cdb2be7/ : SUCCESS in 1h 25m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-dsvm-full-kafka-default/c8313d1/ : SUCCESS in 1h 43m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm/f42f691/ : SUCCESS in 58m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/9/check/oslo.messaging-src-grenade-dsvm-multinode/da80845/ : SUCCESS in 1h 06m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-rabbit/a32f623/ : SUCCESS in 39m 44s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-amqp1/935e8a5/ : FAILURE in 54m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/9/check/oslo.messaging-telemetry-dsvm-integration-kafka/2c09dd5/ : FAILURE in 54m 08s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/e648d41/ : SUCCESS in 1h 17m 53s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/7d7ae18/ : SUCCESS in 1h 46m 30s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/9/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/cfb9e32/ : SUCCESS in 1h 25m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/9/check/openstack-tox-lower-constraints/476900a/ : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":9},{"id":"c58f181111faf5f49edefae46c4da1a859b713e8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-09 17:08:44.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"e74093783d34a3e7da04376fe804d14eb4ff0755","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2018-05-09 17:09:45.000000000","message":"Patch Set 10:\n\n\u003e Ugh - that devstack-amqp1 test is supposed to be non-voting.\n \u003e \n \u003e Can you change the following line in .zuul.yaml:\n \u003e \n \u003e https://git.openstack.org/cgit/openstack/oslo.messaging/tree/.zuul.yaml#n257\n \u003e \n \u003e to be:\n \u003e \n \u003e - oslo.messaging-devstack-amqp1:\n \u003e voting: false\n \u003e \n \u003e I\u0027m ok with piggybacking that change into this patch\n\nSorry, I just saw this. I slipped it into a patch below this. Makes me nervous to make a change and disable a test in the same patch, and I\u0027d normally reject that as a reviewer :)","accounts_in_message":[],"_revision_number":10},{"id":"43ef711371016a8d97932252a7b9f05f9e7968a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-09 20:14:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/10/check/openstack-tox-pep8/f2d5f4a/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/10/check/openstack-tox-py27/955ae01/ : SUCCESS in 6m 41s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/10/check/openstack-tox-py35/2ba6482/ : SUCCESS in 5m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/10/check/build-openstack-sphinx-docs/a875a18/html/ : SUCCESS in 3m 17s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/10/check/legacy-tempest-dsvm-neutron-src/9a024e5/ : SUCCESS in 1h 23m 30s\n- oslo.messaging-devstack-amqp1 http://logs.openstack.org/63/546763/10/check/oslo.messaging-devstack-amqp1/df8409b/ : SUCCESS in 1h 38m 31s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py27-func-amqp1/6f31034/ : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py27-func-kafka/2c603a6/ : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py27-func-rabbit/17a0e39/ : SUCCESS in 4m 44s\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py35-func-amqp1/bcccf8a/ : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py35-func-kafka/b79cdd6/ : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/63/546763/10/check/oslo.messaging-tox-py35-func-rabbit/5658813/ : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-dsvm-full-rabbit-default/e146cf1/ : SUCCESS in 1h 37m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/edea393/ : SUCCESS in 1h 15m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/f0670f8/ : FAILURE in 13m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/2253a5b/ : FAILURE in 12m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-dsvm-full-kafka-default/e68223f/ : SUCCESS in 1h 41m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-grenade-dsvm/099238a/ : SUCCESS in 45m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/63/546763/10/check/oslo.messaging-src-grenade-dsvm-multinode/420dc35/ : SUCCESS in 59m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/10/check/oslo.messaging-telemetry-dsvm-integration-rabbit/144ecf5/ : SUCCESS in 41m 17s\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/63/546763/10/check/oslo.messaging-telemetry-dsvm-integration-amqp1/9b4e39c/ : FAILURE in 48m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/63/546763/10/check/oslo.messaging-telemetry-dsvm-integration-kafka/ed54777/ : FAILURE in 52m 49s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/10/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/a281198/ : SUCCESS in 1h 22m 38s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/63/546763/10/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/a23791e/ : SUCCESS in 1h 16m 52s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/63/546763/10/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/93b8696/ : SUCCESS in 1h 17m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/10/check/openstack-tox-lower-constraints/6bb536a/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":10},{"id":"6d9e739cc87d13a150f015ca32476c77de470102","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2018-05-10 13:42:52.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"04c1dc92c7072e1c4c3f8cb80787cb20e036558d","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2018-05-16 17:51:06.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"387be190f10b9575f75b294ab14533ea4f96add1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 17:51:20.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"58e1efe6df3c5eaa95c0abf1f9d17e1988426854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 19:25:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"ab622c4be5bfa8e64453ed4a6a392b9c2298a210","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 19:25:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/63/546763/10/gate/openstack-tox-pep8/991ef35/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/63/546763/10/gate/openstack-tox-py27/fec4b32/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/63/546763/10/gate/openstack-tox-py35/11c51f2/ : SUCCESS in 3m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/546763/10/gate/build-openstack-sphinx-docs/c063979/html/ : SUCCESS in 2m 43s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/63/546763/10/gate/legacy-tempest-dsvm-neutron-src/f05f898/ : SUCCESS in 1h 31m 57s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/63/546763/10/gate/oslo.messaging-tox-py27-func-rabbit/bf96975/ : SUCCESS in 3m 56s\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/63/546763/10/gate/oslo.messaging-telemetry-dsvm-integration-rabbit/51300f7/ : SUCCESS in 42m 05s\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/63/546763/10/gate/oslo.messaging-src-dsvm-full-rabbit-default/fa05dd4/ : SUCCESS in 1h 32m 35s\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/63/546763/10/gate/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/4aa05fd/ : SUCCESS in 1h 15m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/546763/10/gate/openstack-tox-lower-constraints/e9e8fc3/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b34ab8b1cc9f4d513a2927c102dbbe82031d9c2a","revisions":{"8458e796b7678f46133684c8f2b222f5db06f3b2":{"kind":"REWORK","_number":1,"created":"2018-02-21 20:17:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/1"}}},"commit":{"parents":[{"commit":"6bdae936581b9d8acd6ab28809e6a40297b16ae0","subject":"Add rabbitmq-server for platform:rpm into bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bdae936581b9d8acd6ab28809e6a40297b16ae0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 20:17:02.000000000","tz":-480},"subject":"WIP: Heartbeat active calls","message":"WIP: Heartbeat active calls\n\nCompletely hacky local PoC for this thing. Don\u0027t review it, I\u0027m just putting this\nup to collaborate.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8458e796b7678f46133684c8f2b222f5db06f3b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8458e796b7678f46133684c8f2b222f5db06f3b2"}]},"branch":"refs/heads/master"},"5a0b16f475fd96380ad407a080d1afe3b91a4134":{"kind":"REWORK","_number":2,"created":"2018-03-03 22:26:12.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/2"}}},"commit":{"parents":[{"commit":"b1d651fb4ac9f7b535aecb7cf54f95e059a76161","subject":"Add heartbeat() method to RpcIncomingMessage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b1d651fb4ac9f7b535aecb7cf54f95e059a76161"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-03-03 22:26:03.000000000","tz":-480},"subject":"WIP: Heartbeat active calls","message":"WIP: Heartbeat active calls\n\nStill work to do on this, but a little closer and I want to get a test run with\nsome of the recent updates.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a0b16f475fd96380ad407a080d1afe3b91a4134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5a0b16f475fd96380ad407a080d1afe3b91a4134"}]},"branch":"refs/heads/master"},"89f502dd35865abfa8aa9214083bff63ebb30954":{"kind":"REWORK","_number":3,"created":"2018-03-07 20:17:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/3"}}},"commit":{"parents":[{"commit":"3ddef68360b183745199ed492ffbc507adddf6a3","subject":"Add heartbeat() method to RpcIncomingMessage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ddef68360b183745199ed492ffbc507adddf6a3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-03-07 20:16:13.000000000","tz":-480},"subject":"WIP: Heartbeat active calls","message":"WIP: Heartbeat active calls\n\nSome things to figure out yet:\n\n - Defaulting heartbeat_timeout to timeout/30\n - Allow disabling heartbeat_timeout some way\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89f502dd35865abfa8aa9214083bff63ebb30954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89f502dd35865abfa8aa9214083bff63ebb30954"}]},"branch":"refs/heads/master"},"52c78ee0e4c014041b64237b85ecc25b792ff249":{"kind":"REWORK","_number":4,"created":"2018-03-08 19:51:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/4"}}},"commit":{"parents":[{"commit":"3ddef68360b183745199ed492ffbc507adddf6a3","subject":"Add heartbeat() method to RpcIncomingMessage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ddef68360b183745199ed492ffbc507adddf6a3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-03-08 19:50:35.000000000","tz":-480},"subject":"WIP: Heartbeat active calls","message":"WIP: Heartbeat active calls\n\nSome things to figure out yet:\n\n - Defaulting heartbeat_timeout to timeout/30\n - Allow disabling heartbeat_timeout some way\n - Wait the full timeout time for the first heartbeat\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/52c78ee0e4c014041b64237b85ecc25b792ff249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/52c78ee0e4c014041b64237b85ecc25b792ff249"}]},"branch":"refs/heads/master"},"aa8be93849c764ff3b25bda5cfc68ff3836eb006":{"kind":"REWORK","_number":5,"created":"2018-04-24 16:51:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/5"}}},"commit":{"parents":[{"commit":"930e6189e25425576ed3a2f570bda3fd9d310bd6","subject":"Add heartbeat() method to RpcIncomingMessage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/930e6189e25425576ed3a2f570bda3fd9d310bd6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-04-24 16:42:21.000000000","tz":-420},"subject":"[rabbitmq] Heartbeat active calls","message":"[rabbitmq] Heartbeat active calls\n\nThis adds an optional heartbeat_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the heartbeat_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aa8be93849c764ff3b25bda5cfc68ff3836eb006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aa8be93849c764ff3b25bda5cfc68ff3836eb006"}]},"branch":"refs/heads/master"},"3f39d325df3bf6ec6d7c88f10403e6dd9668e343":{"kind":"REWORK","_number":6,"created":"2018-05-02 16:50:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/6"}}},"commit":{"parents":[{"commit":"930e6189e25425576ed3a2f570bda3fd9d310bd6","subject":"Add heartbeat() method to RpcIncomingMessage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/930e6189e25425576ed3a2f570bda3fd9d310bd6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-05-02 16:49:59.000000000","tz":-420},"subject":"[rabbitmq] Heartbeat active calls","message":"[rabbitmq] Heartbeat active calls\n\nThis adds an optional heartbeat_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the heartbeat_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3f39d325df3bf6ec6d7c88f10403e6dd9668e343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3f39d325df3bf6ec6d7c88f10403e6dd9668e343"}]},"branch":"refs/heads/master"},"5f184be8119d5d5f96d66158e9acae7d1418bf57":{"kind":"REWORK","_number":7,"created":"2018-05-03 16:28:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/7"}}},"commit":{"parents":[{"commit":"8214e924fc1d20bbf07df1eee8419a6c663dc1ed","subject":"Merge \"Remove stale pip-missing-reqs tox test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8214e924fc1d20bbf07df1eee8419a6c663dc1ed"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-05-03 16:28:47.000000000","tz":-420},"subject":"[rabbitmq] Heartbeat active calls","message":"[rabbitmq] Heartbeat active calls\n\nThis adds an optional heartbeat_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the heartbeat_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f184be8119d5d5f96d66158e9acae7d1418bf57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f184be8119d5d5f96d66158e9acae7d1418bf57"}]},"branch":"refs/heads/master"},"7f9eaab08a22fd784c8f280b03f139a3b81d278e":{"kind":"REWORK","_number":8,"created":"2018-05-03 16:39:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/8"}}},"commit":{"parents":[{"commit":"8214e924fc1d20bbf07df1eee8419a6c663dc1ed","subject":"Merge \"Remove stale pip-missing-reqs tox test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8214e924fc1d20bbf07df1eee8419a6c663dc1ed"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-05-03 16:39:37.000000000","tz":-420},"subject":"[rabbitmq] Heartbeat active calls","message":"[rabbitmq] Heartbeat active calls\n\nThis adds an optional heartbeat_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the heartbeat_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f9eaab08a22fd784c8f280b03f139a3b81d278e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f9eaab08a22fd784c8f280b03f139a3b81d278e"}]},"branch":"refs/heads/master"},"0fa3bec06ec57ff262d5d2a87fde892235c1f1d2":{"kind":"REWORK","_number":9,"created":"2018-05-08 17:09:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/9"}}},"commit":{"parents":[{"commit":"8214e924fc1d20bbf07df1eee8419a6c663dc1ed","subject":"Merge \"Remove stale pip-missing-reqs tox test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8214e924fc1d20bbf07df1eee8419a6c663dc1ed"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-05-08 17:08:09.000000000","tz":-420},"subject":"[rabbitmq] Implement active call monitoring","message":"[rabbitmq] Implement active call monitoring\n\nThis adds an optional call_monitor_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the call_monitor_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0fa3bec06ec57ff262d5d2a87fde892235c1f1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0fa3bec06ec57ff262d5d2a87fde892235c1f1d2"}]},"branch":"refs/heads/master"},"b34ab8b1cc9f4d513a2927c102dbbe82031d9c2a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2018-05-09 17:08:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/546763/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/546763/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/546763/10"}}},"commit":{"parents":[{"commit":"d38ad3e4656e0f2db217467ea7fcfc6d0e3e7202","subject":"Make oslo.messaging-devstack-amqp1 job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d38ad3e4656e0f2db217467ea7fcfc6d0e3e7202"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-02-21 19:03:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2018-05-09 17:08:37.000000000","tz":-420},"subject":"[rabbitmq] Implement active call monitoring","message":"[rabbitmq] Implement active call monitoring\n\nThis adds an optional call_monitor_timeout parameter to the RPC client,\nwhich if specified, will enable heartbeating of long-running calls by\nthe server. This enables the user to increase the regular timeout to\na much larger value, allowing calls to take a very long time, but\nwith heartbeating to indicate that they are still running on the server\nside. If the server stops heartbeating, then the call_monitor_timeout\ntakes over and we fail with the usual MessagingTimeout instead of waiting\nfor the longer overall timeout to expire.\n\nChange-Id: I60334aaf019f177a984583528b71d00859d31f84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b34ab8b1cc9f4d513a2927c102dbbe82031d9c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b34ab8b1cc9f4d513a2927c102dbbe82031d9c2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
