)]}'
{"id":"openstack%2Foslo.messaging~649989","triplet_id":"openstack%2Foslo.messaging~master~I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1822778","hashtags":[],"change_id":"I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","subject":"Retry to declare a queue after internal error","status":"MERGED","created":"2019-04-04 13:20:31.000000000","updated":"2019-04-23 07:51:00.000000000","submitted":"2019-04-22 22:23:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"649989-1555971787557-5004bf0a","meta_rev_id":"9f7efc318d1bc038ccd903a3d746756133760598","_number":649989,"virtual_id_number":649989,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2019-04-22 22:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-04-15 12:32:18.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":1,"date":"2019-04-08 17:51:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-04-22 20:19:42.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":1,"date":"2019-04-22 20:19:42.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-08 17:51:02.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-04-15 12:32:18.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-04-22 20:19:42.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-04-22 22:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a33eab16309bcccbde1182fd47e6dc078180ca9c","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-04 13:20:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e7acd914f1f4416097a788d78e18e4c4a8f8253","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-04 13:46:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"844f071b47f75365eddeba093219e54cc8b86ad6","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-04 14:12:10.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d85f352cadb36850ec10d8d99811ebd770e33db4","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-04-04 14:17:54.000000000","message":"Patch Set 1:\n\n(5 comments)\n\n@Ken how do you handle this situation when you add a new option to fix a bug. is this backportable? should it be mentioned in the release notes?","accounts_in_message":[],"_revision_number":1},{"id":"31579561bb62115c4bab21d954b3f1aec8e2f8f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 15:52:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/89/649989/1/check/tempest-full/e62dedf/ : TIMED_OUT in 2h 19m 49s\n- tempest-full-py3 http://logs.openstack.org/89/649989/1/check/tempest-full-py3/cbda7a6/ : SUCCESS in 1h 35m 00s\n- openstack-tox-cover http://logs.openstack.org/89/649989/1/check/openstack-tox-cover/758faa1/cover/ : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/649989/1/check/openstack-tox-lower-constraints/3663913/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 http://logs.openstack.org/89/649989/1/check/openstack-tox-pep8/19e7a03/ : FAILURE in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/89/649989/1/check/openstack-tox-py27/1a4c536/ : SUCCESS in 5m 58s\n- openstack-tox-py36 http://logs.openstack.org/89/649989/1/check/openstack-tox-py36/2c5af1a/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/89/649989/1/check/openstack-tox-py37/b0bc8b4/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/89/649989/1/check/openstack-tox-docs/c67425e/html/ : SUCCESS in 3m 42s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py27-func-rabbit/15aef4a/ : SUCCESS in 5m 07s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py27-func-amqp1/53d928d/ : SUCCESS in 4m 46s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py27-func-kafka/7ed1584/ : FAILURE in 4m 28s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py35-func-rabbit/969d6d1/ : FAILURE in 9m 58s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py35-func-amqp1/0486ada/ : SUCCESS in 6m 06s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/89/649989/1/check/oslo.messaging-tox-py35-func-kafka/d5d89f8/ : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-dsvm-full-rabbit/434e5b4/ : SUCCESS in 1h 29m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/6dfd6d4/ : SUCCESS in 2h 03m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/712ab36/ : SUCCESS in 2h 06m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/872ebca/ : FAILURE in 18m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/c3e8c2c/ : FAILURE in 21m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-grenade-dsvm/9fcb8bb/ : SUCCESS in 1h 51m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/89/649989/1/check/oslo.messaging-src-grenade-dsvm-multinode/7c26c9d/ : SUCCESS in 1h 25m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/89/649989/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/2971b28/ : FAILURE in 2h 28m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/89/649989/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/71ac3ce/ : FAILURE in 1h 06m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/89/649989/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/4ac535d/ : FAILURE in 48m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"692e32d5f24a2277bea577ef3af9053d0e34577f","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-04 17:16:35.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThe additional error handling is fine, I\u0027d just avoid adding a whole new configuration attribute for it since its a simple workaround to what appears as a somewhat infrequent cluster timing bug.\n\n@Antonio - if we drop the config attribute we can probably avoid a release note as the launchpad bug is associated with it.","accounts_in_message":[],"_revision_number":1},{"id":"ba8c2ecd8e5fdc4a79a1802ba927ba1c0f35c9e3","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-05 13:06:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"988160f394b464a8343237cc601dfef7b0f6d8ad","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-05 13:08:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f336a621114779e035bcc9adff339f1e9351125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-05 15:11:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/89/649989/3/check/tempest-full/ac444b4/ : SUCCESS in 1h 44m 13s\n- tempest-full-py3 http://logs.openstack.org/89/649989/3/check/tempest-full-py3/142daa7/ : SUCCESS in 1h 38m 12s\n- openstack-tox-cover http://logs.openstack.org/89/649989/3/check/openstack-tox-cover/eccf0fc/cover/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/649989/3/check/openstack-tox-lower-constraints/4632817/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/89/649989/3/check/openstack-tox-pep8/08a5a0a/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/89/649989/3/check/openstack-tox-py27/765b442/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/89/649989/3/check/openstack-tox-py36/eb4527c/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/89/649989/3/check/openstack-tox-py37/076ae6a/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/89/649989/3/check/openstack-tox-docs/cf434d2/html/ : SUCCESS in 3m 33s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-rabbit/5a0cf64/ : SUCCESS in 5m 18s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-amqp1/866c33b/ : SUCCESS in 4m 22s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-kafka/2edc6e1/ : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-rabbit/464ea36/ : SUCCESS in 5m 50s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-amqp1/a371cb3/ : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-kafka/adad76c/ : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-rabbit/85e8cb9/ : FAILURE in 1h 26m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/51df931/ : SUCCESS in 1h 38m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/89fd564/ : SUCCESS in 2h 00m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/5697085/ : FAILURE in 18m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/19aeac5/ : FAILURE in 25m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-grenade-dsvm/ba511fd/ : SUCCESS in 52m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-grenade-dsvm-multinode/6bc5831/ : SUCCESS in 1h 08m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/ad39811/ : FAILURE in 1h 06m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/4aece50/ : FAILURE in 1h 05m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/c80906b/ : FAILURE in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23f78a382905567cd6faf74d511fbdfe4b18e61a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-08 13:38:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"38800a2026b4a09239e5ef47104d901805ba094b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 15:41:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/89/649989/3/check/tempest-full/c79696a/ : SUCCESS in 1h 50m 31s\n- tempest-full-py3 http://logs.openstack.org/89/649989/3/check/tempest-full-py3/312494e/ : SUCCESS in 1h 37m 09s\n- openstack-tox-cover http://logs.openstack.org/89/649989/3/check/openstack-tox-cover/9e8e5cf/cover/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/649989/3/check/openstack-tox-lower-constraints/fb61703/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/89/649989/3/check/openstack-tox-pep8/0328bd1/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/89/649989/3/check/openstack-tox-py27/99f7c65/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/89/649989/3/check/openstack-tox-py36/01fbe9c/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/89/649989/3/check/openstack-tox-py37/b279ec0/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/89/649989/3/check/openstack-tox-docs/62dad0a/html/ : SUCCESS in 3m 45s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-rabbit/f35e576/ : SUCCESS in 5m 44s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-amqp1/62b764a/ : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py27-func-kafka/e0e7406/ : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-rabbit/992980e/ : SUCCESS in 5m 32s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-amqp1/7c9d0c0/ : SUCCESS in 4m 47s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-tox-py35-func-kafka/e8d6d88/ : FAILURE in 5m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-rabbit/9a42b6c/ : SUCCESS in 1h 37m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/0ea6479/ : SUCCESS in 1h 36m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/893f7c4/ : SUCCESS in 1h 56m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3ae0823/ : FAILURE in 23m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/ebc5e8e/ : FAILURE in 20m 46s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-grenade-dsvm/f40e0d4/ : SUCCESS in 1h 01m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/89/649989/3/check/oslo.messaging-src-grenade-dsvm-multinode/60dcdfe/ : SUCCESS in 1h 05m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/bb1b4a0/ : FAILURE in 1h 10m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/03ec96c/ : FAILURE in 1h 05m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/89/649989/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/afbdc46/ : FAILURE in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"06d9150625753f463bd57b9bdec3739a7c67e7b3","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-04-08 17:51:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5db7ca5a5744918ae98222eed08410c0da78f86e","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-15 12:32:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ba85fa1b31d2c9fa26014eb00598eed66395fbcb","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-04-22 20:19:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dde3426a23490d1b3f4875f5cdf0dc3784d2a284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:19:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d72032feab092b7a471dd45904af604022fcdbe","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-04-22 20:20:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ff17a611d73c3041e8aaffdd68cbe5e1c9ae576a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:23:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/89/649989/3/gate/tempest-full/c0ed361/ : SUCCESS in 1h 57m 31s\n- tempest-full-py3 http://logs.openstack.org/89/649989/3/gate/tempest-full-py3/d8ff3cd/ : SUCCESS in 1h 33m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/649989/3/gate/openstack-tox-lower-constraints/aae6a3f/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/89/649989/3/gate/openstack-tox-pep8/7412b0c/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/89/649989/3/gate/openstack-tox-py27/bccac9f/ : SUCCESS in 6m 31s\n- openstack-tox-py36 http://logs.openstack.org/89/649989/3/gate/openstack-tox-py36/9a7730c/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/89/649989/3/gate/openstack-tox-py37/40dae1e/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/89/649989/3/gate/openstack-tox-docs/5dcc75b/html/ : SUCCESS in 3m 28s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/649989/3/gate/oslo.messaging-tox-py27-func-rabbit/2db4e00/ : SUCCESS in 5m 48s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/649989/3/gate/oslo.messaging-src-dsvm-full-rabbit/7da9bee/ : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":3},{"id":"39cfcca810437d6ad1a30bb810bcde6cb31df5e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:23:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"faaa2a01dce4c2920cf99e9d6c7fda2825c15cca","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:49:37.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 0297c5f0ccfa411aded1a1cb3559de90aa276dd3","accounts_in_message":[],"_revision_number":3},{"id":"7bce2e508a144307e1cb6263f21b3e051c8c0410","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:50:24.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 6ea1bb5fad84079d07d8ad48d2c2a71aa29cdda4","accounts_in_message":[],"_revision_number":3},{"id":"b02eaa42f35a070ce3e14052cf548ab5cf1462de","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:50:41.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit d75eba02c74d309cd769a8257d2d20f735c60ad1","accounts_in_message":[],"_revision_number":3},{"id":"91e52396ce2be989158d4b85e6181e027d65f993","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:51:00.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 99b77c07bee7dc482af03d13b2c454593f153f0c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d2787227b00b973973554f7387e621d2664c0d8","revisions":{"36c3690792c668d988129897eaf31f00c83cb051":{"kind":"REWORK","_number":1,"created":"2019-04-04 13:20:31.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/89/649989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/649989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/1"}}},"commit":{"parents":[{"commit":"44bd971bfb134cd74d09322a6bfc7449e3be577e","subject":"Unmark RabbitMQ heartbeat as experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44bd971bfb134cd74d09322a6bfc7449e3be577e"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 13:20:28.000000000","tz":120},"subject":"WIP Retry to declare a queue after internal error","message":"WIP Retry to declare a queue after internal error\n\nCloses-Bug: [1]\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/36c3690792c668d988129897eaf31f00c83cb051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/36c3690792c668d988129897eaf31f00c83cb051"}]},"branch":"refs/heads/master"},"3e2a1283971a43c4e561de9873a9b0b447452b19":{"kind":"REWORK","_number":2,"created":"2019-04-05 13:06:52.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/89/649989/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/649989/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/2"}}},"commit":{"parents":[{"commit":"44bd971bfb134cd74d09322a6bfc7449e3be577e","subject":"Unmark RabbitMQ heartbeat as experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44bd971bfb134cd74d09322a6bfc7449e3be577e"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-05 13:05:26.000000000","tz":120},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3e2a1283971a43c4e561de9873a9b0b447452b19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3e2a1283971a43c4e561de9873a9b0b447452b19"}]},"branch":"refs/heads/master"},"4d2787227b00b973973554f7387e621d2664c0d8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-04-05 13:08:39.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/89/649989/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/649989/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/649989/3"}}},"commit":{"parents":[{"commit":"44bd971bfb134cd74d09322a6bfc7449e3be577e","subject":"Unmark RabbitMQ heartbeat as experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44bd971bfb134cd74d09322a6bfc7449e3be577e"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-05 13:08:12.000000000","tz":120},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: #1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d2787227b00b973973554f7387e621d2664c0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d2787227b00b973973554f7387e621d2664c0d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
