)]}'
{"id":"openstack%2Foslo.messaging~651765","triplet_id":"openstack%2Foslo.messaging~master~I438c7279d584bdb00da18477a83d5e2a53c588c4","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1824339","hashtags":[],"change_id":"I438c7279d584bdb00da18477a83d5e2a53c588c4","subject":"Consider the topic parameter as an array in client-notify","status":"MERGED","created":"2019-04-11 11:55:22.000000000","updated":"2019-04-29 22:36:42.000000000","submitted":"2019-04-29 22:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651765-1556577402716-6fb4be29","meta_rev_id":"e3b57299e106ed4463707571152cc113f78067d4","_number":651765,"virtual_id_number":651765,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2019-04-11 17:08:49.000000000","_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2019-04-29 22:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-04-15 12:31:26.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":1,"date":"2019-04-12 07:00:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-04-29 20:31:14.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":1,"date":"2019-04-29 20:31:14.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-12 07:00:41.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-04-15 12:31:26.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-04-29 20:31:14.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-04-29 22:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4f88ec8d67e508643d094fe1eae3956df1ce082","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-11 11:55:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f141d45815c678a6027e6f9b6d22ee0eb28364c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 14:20:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/65/651765/1/check/tempest-full/094284e/ : SUCCESS in 1h 47m 52s\n- tempest-full-py3 http://logs.openstack.org/65/651765/1/check/tempest-full-py3/0432f39/ : SUCCESS in 1h 21m 54s\n- openstack-tox-cover http://logs.openstack.org/65/651765/1/check/openstack-tox-cover/95f5c71/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/651765/1/check/openstack-tox-lower-constraints/cd77e81/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/65/651765/1/check/openstack-tox-pep8/572ffbe/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/65/651765/1/check/openstack-tox-py27/5730508/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/65/651765/1/check/openstack-tox-py36/cbdebaa/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/65/651765/1/check/openstack-tox-py37/2265e09/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/65/651765/1/check/openstack-tox-docs/74d2475/html/ : SUCCESS in 2m 51s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-rabbit/946de01/ : FAILURE in 5m 59s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-amqp1/a239c81/ : SUCCESS in 4m 31s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-kafka/8938c2e/ : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-rabbit/706232e/ : SUCCESS in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-amqp1/27b5508/ : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-kafka/7551100/ : FAILURE in 4m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-rabbit/3aef3bd/ : SUCCESS in 1h 31m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/0831d50/ : SUCCESS in 1h 36m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/c2104f9/ : SUCCESS in 2h 13m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3e6ef16/ : FAILURE in 19m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/0b4e738/ : FAILURE in 23m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-grenade-dsvm/42189af/ : SUCCESS in 1h 00m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-grenade-dsvm-multinode/5fc2193/ : SUCCESS in 1h 05m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/f5c6af2/ : FAILURE in 1h 17m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/dce97df/ : FAILURE in 1h 03m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/312c326/ : FAILURE in 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a5b4171dfc02a3a5f98aca23c44cc5d48699f3c","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-04-11 14:32:06.000000000","message":"Patch Set 1:\n\nare those job failures related to this change?","accounts_in_message":[],"_revision_number":1},{"id":"f743ced9e6b5210fc353806c25de6bb31c2a56c5","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-11 17:08:49.000000000","message":"Patch Set 1:\n\n@antonio, \nI don\u0027t think so, it fails here:\nhttp://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-rabbit/946de01/job-output.txt.gz#_2019-04-11_12_03_12_563067\nI just changed the simulator","accounts_in_message":[],"_revision_number":1},{"id":"a473c8ba584eff00d81862f4dea92707b0ae754f","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-04-11 20:40:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7394cdae2d2f754bdb1ddfd20fac32cdb5f30af8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 23:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/65/651765/1/check/tempest-full/704f951/ : SUCCESS in 2h 01m 21s\n- tempest-full-py3 http://logs.openstack.org/65/651765/1/check/tempest-full-py3/3c6b9f6/ : SUCCESS in 1h 40m 15s\n- openstack-tox-cover http://logs.openstack.org/65/651765/1/check/openstack-tox-cover/545b477/cover/ : SUCCESS in 7m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/651765/1/check/openstack-tox-lower-constraints/bb808ec/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/65/651765/1/check/openstack-tox-pep8/12377a9/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/65/651765/1/check/openstack-tox-py27/365b3f2/ : SUCCESS in 7m 28s\n- openstack-tox-py36 http://logs.openstack.org/65/651765/1/check/openstack-tox-py36/39656bf/ : SUCCESS in 7m 27s\n- openstack-tox-py37 http://logs.openstack.org/65/651765/1/check/openstack-tox-py37/95d4647/ : SUCCESS in 6m 55s\n- openstack-tox-docs http://logs.openstack.org/65/651765/1/check/openstack-tox-docs/0df1210/html/ : SUCCESS in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-rabbit/277fcb5/ : SUCCESS in 8m 03s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-amqp1/eed384b/ : SUCCESS in 6m 34s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py27-func-kafka/b599f53/ : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-rabbit/83b846a/ : SUCCESS in 6m 50s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-amqp1/9e5b0c0/ : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-tox-py35-func-kafka/054564f/ : FAILURE in 6m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-rabbit/3ba2e88/ : SUCCESS in 1h 43m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/f0e8d90/ : SUCCESS in 1h 43m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/6619f95/ : SUCCESS in 2h 11m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/04c0265/ : FAILURE in 22m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/c13f959/ : FAILURE in 29m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-grenade-dsvm/09fc229/ : SUCCESS in 1h 23m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/65/651765/1/check/oslo.messaging-src-grenade-dsvm-multinode/725e483/ : SUCCESS in 1h 17m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/d3725c7/ : FAILURE in 1h 10m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c4ef553/ : FAILURE in 1h 03m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/65/651765/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/d9df3ed/ : FAILURE in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ada2300b35808a3f9912a91058d66ec763868c7","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-04-12 07:00:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"50ce2aa79bf914555831e926d413737abe8a69d1","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-15 12:31:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"790bed971815332cdc90b3f43f9063af4ec5f137","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-04-29 20:31:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56bd002ea3b4120f44a172f939796bf352cd09d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 20:31:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eabc6d8d0cccae95c80a98aa27b7922233f596bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 22:36:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4977b72258dcda25cf6dd024057556e6cdc788d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 22:36:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/65/651765/1/gate/tempest-full/8568df2/ : SUCCESS in 1h 54m 13s\n- tempest-full-py3 http://logs.openstack.org/65/651765/1/gate/tempest-full-py3/cf8eaca/ : SUCCESS in 1h 35m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/651765/1/gate/openstack-tox-lower-constraints/34b6b34/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/65/651765/1/gate/openstack-tox-pep8/3f5b763/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/65/651765/1/gate/openstack-tox-py27/3a549f9/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/65/651765/1/gate/openstack-tox-py36/43770d5/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/65/651765/1/gate/openstack-tox-py37/2689d8e/ : SUCCESS in 5m 37s\n- openstack-tox-docs http://logs.openstack.org/65/651765/1/gate/openstack-tox-docs/f191efe/html/ : SUCCESS in 2m 54s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/65/651765/1/gate/oslo.messaging-tox-py27-func-rabbit/99543a9/ : SUCCESS in 5m 40s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/65/651765/1/gate/oslo.messaging-src-dsvm-full-rabbit/b8db510/ : SUCCESS in 1h 48m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ad4a204b385273b671327028d30133725c6174e","revisions":{"0ad4a204b385273b671327028d30133725c6174e":{"kind":"REWORK","_number":1,"created":"2019-04-11 11:55:22.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/65/651765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/651765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/651765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/651765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/651765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/651765/1"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-11 11:48:40.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-11 11:55:21.000000000","tz":120},"subject":"Consider the topic parameter as an array in client-notify","message":"Consider the topic parameter as an array in client-notify\n\nWithout this fix, the simulator splits the \"topic\" value in different\nstrings, in notify-client case\nInstead of having one queue there are multi-queues.\nSee [1] for more details\n\nCloses-Bug: #1824339\n\n1- https://bugs.launchpad.net/oslo.messaging/+bug/1824339\n\nChange-Id: I438c7279d584bdb00da18477a83d5e2a53c588c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ad4a204b385273b671327028d30133725c6174e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ad4a204b385273b671327028d30133725c6174e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
