)]}'
{"id":"openstack%2Foslo.messaging~652056","triplet_id":"openstack%2Foslo.messaging~master~I9ddb6595fc52e46ed8844e39d2fa71029c90b65c","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1674567","hashtags":[],"change_id":"I9ddb6595fc52e46ed8844e39d2fa71029c90b65c","subject":"Remove log translation and i18n","status":"MERGED","created":"2019-04-12 12:35:48.000000000","updated":"2019-05-04 22:31:13.000000000","submitted":"2019-05-04 22:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"652056-1557009073490-22df3888","meta_rev_id":"bbd2de72a09c592081ed65aa6cdcbe761c25a5e0","_number":652056,"virtual_id_number":652056,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-04 22:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-05-04 20:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-04-24 13:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-05-04 20:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 14:45:19.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-04-23 19:50:26.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-04-24 13:17:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-04 20:23:01.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-05-04 22:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92731f4a77a9a087cdbc9e4daafb34d5ca8535b0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-12 12:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8454d8ae8cb65bc382c517318e2b13619e010c7","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-12 12:42:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8368580735bdb4dbd9915ec03e374d30d1ea1c4a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-12 12:46:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef3435b7edd6a9b3ee6436345b986c16ddcafb97","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-12 13:18:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9feb810e0263ba7c099db3cec10c2573efea854f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-12 13:21:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nThanks for doing this :) I have the same comments for all files:\n\n- Don\u0027t use backslashes for line wrapping\n- Don\u0027t use string interpolation for logging\n\nFix these and I\u0027m +2 :)","accounts_in_message":[],"_revision_number":2},{"id":"ed61048776cf3e58830551f77e55f4e59c1fd2e9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-12 13:56:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffb87b3181f2e05a64f2ac9999944cc63d7b90d9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-12 13:56:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6bc4395d83b0634f827675191ddd27526de8c64d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-12 14:41:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(15 comments)\n\nStill some left over. Not sure about not translating exceptions (there are more than one, sadly). Who can we ask about this?","accounts_in_message":[],"_revision_number":3},{"id":"b5132c93bc5547b36053db98e89fc609ccb9b172","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 16:23:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/56/652056/3/check/requirements-check/1274163/ : SUCCESS in 4m 58s\n- tempest-full http://logs.openstack.org/56/652056/3/check/tempest-full/8d4a94d/ : SUCCESS in 1h 57m 41s\n- tempest-full-py3 http://logs.openstack.org/56/652056/3/check/tempest-full-py3/1c3d258/ : SUCCESS in 1h 46m 25s\n- openstack-tox-cover http://logs.openstack.org/56/652056/3/check/openstack-tox-cover/91e28bb/cover/ : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/3/check/openstack-tox-lower-constraints/5c10af7/ : SUCCESS in 7m 52s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/3/check/openstack-tox-pep8/8332fae/ : SUCCESS in 7m 38s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/3/check/openstack-tox-py27/ef54349/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/3/check/openstack-tox-py36/3f85e1c/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/3/check/openstack-tox-py37/ad2ab9f/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/56/652056/3/check/openstack-tox-docs/dc027f2/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/3/check/build-openstack-releasenotes/ffff7c3/html/ : SUCCESS in 3m 45s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py27-func-rabbit/9741e71/ : SUCCESS in 5m 21s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py27-func-amqp1/fc60e70/ : SUCCESS in 5m 09s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py27-func-kafka/32410a4/ : FAILURE in 4m 56s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py35-func-rabbit/2716244/ : SUCCESS in 5m 25s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py35-func-amqp1/83225f6/ : SUCCESS in 5m 53s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/56/652056/3/check/oslo.messaging-tox-py35-func-kafka/eeda691/ : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-dsvm-full-rabbit/a26a541/ : SUCCESS in 2h 20m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/ce819dc/ : SUCCESS in 1h 48m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/74a4fef/ : FAILURE in 22m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/f95fddd/ : FAILURE in 16m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/1ef0e04/ : FAILURE in 29m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-grenade-dsvm/81b4396/ : SUCCESS in 1h 01m 43s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/3/check/oslo.messaging-src-grenade-dsvm-multinode/309a4f6/ : SUCCESS in 1h 15m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/df8baca/ : FAILURE in 57m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/4d44978/ : FAILURE in 1h 06m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/fcdf7f6/ : FAILURE in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"916928b122ed15b105eaabc08938567ec6eb9421","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-15 10:13:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"105c93da4c8677a3f472e5698027ccc9d01f5f47","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-15 10:14:18.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"182e42cc1f71a861492d00311d7dfed923174b3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 12:32:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/652056/4/check/requirements-check/98e41f8/ : SUCCESS in 3m 04s\n- tempest-full http://logs.openstack.org/56/652056/4/check/tempest-full/e5df80c/ : TIMED_OUT in 2h 05m 37s\n- tempest-full-py3 http://logs.openstack.org/56/652056/4/check/tempest-full-py3/186b437/ : FAILURE in 45m 44s\n- openstack-tox-cover http://logs.openstack.org/56/652056/4/check/openstack-tox-cover/7289b9b/cover/ : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/4/check/openstack-tox-lower-constraints/7cad445/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/4/check/openstack-tox-pep8/8d9ee71/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/4/check/openstack-tox-py27/e3edbbb/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/4/check/openstack-tox-py36/4a6c4ab/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/4/check/openstack-tox-py37/a229a95/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/56/652056/4/check/openstack-tox-docs/39dcbe9/html/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/4/check/build-openstack-releasenotes/389fed7/html/ : SUCCESS in 3m 48s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-rabbit/d0bc722/ : FAILURE in 5m 41s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-amqp1/8ff9cc1/ : FAILURE in 5m 04s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-kafka/9b11389/ : FAILURE in 5m 14s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-rabbit/607196f/ : FAILURE in 7m 07s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-amqp1/bea93be/ : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-kafka/96ecc47/ : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-rabbit/b74fcb6/ : SUCCESS in 1h 31m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/9e2bdbf/ : SUCCESS in 1h 37m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/bc6996b/ : SUCCESS in 2h 12m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-hybrid/9e7c50c/ : FAILURE in 21m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-centos-7/58851a7/ : FAILURE in 24m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm/dfe9cc4/ : SUCCESS in 1h 00m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm-multinode/617872f/ : SUCCESS in 1h 11m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/7899130/ : FAILURE in 1h 06m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/1375971/ : FAILURE in 56m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/a612016/ : FAILURE in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03b10629268ad355a978fb5fd15861f2ae7e8a84","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-15 13:19:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8d33d1a4d8936b99c4babce811e7dca4e1174726","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 15:44:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/652056/4/check/requirements-check/3b2f53f/ : SUCCESS in 3m 22s\n- tempest-full http://logs.openstack.org/56/652056/4/check/tempest-full/46e0444/ : SUCCESS in 1h 56m 05s\n- tempest-full-py3 http://logs.openstack.org/56/652056/4/check/tempest-full-py3/079ba62/ : SUCCESS in 1h 25m 28s\n- openstack-tox-cover http://logs.openstack.org/56/652056/4/check/openstack-tox-cover/e543742/cover/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/4/check/openstack-tox-lower-constraints/cd62e56/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/4/check/openstack-tox-pep8/d48087e/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/4/check/openstack-tox-py27/1e5a54e/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/4/check/openstack-tox-py36/c55208e/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/4/check/openstack-tox-py37/5090fc3/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/56/652056/4/check/openstack-tox-docs/3e8847c/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/4/check/build-openstack-releasenotes/cf65ee8/html/ : SUCCESS in 4m 15s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-rabbit/20331ef/ : FAILURE in 5m 09s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-amqp1/2af369c/ : SUCCESS in 4m 13s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-kafka/6d793ad/ : FAILURE in 6m 29s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-rabbit/5ece9e4/ : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-amqp1/fee1f3c/ : SUCCESS in 4m 56s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-kafka/3a762fa/ : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-rabbit/165c3e0/ : SUCCESS in 1h 43m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/6c84426/ : SUCCESS in 1h 38m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/38dd8c1/ : SUCCESS in 2h 00m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-hybrid/35613bf/ : FAILURE in 19m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-centos-7/b213806/ : FAILURE in 27m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm/3d3ff28/ : SUCCESS in 1h 04m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm-multinode/9a1d5ff/ : SUCCESS in 1h 07m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/1160818/ : FAILURE in 1h 00m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c69d9c9/ : FAILURE in 1h 06m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/316e02b/ : FAILURE in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d74b9ce74360a54812f34b043217b245c03ef8b5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-16 08:15:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"678da0251c30094aa1e599eb62b55a2e462653df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 10:38:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/652056/4/check/requirements-check/78eeda3/ : SUCCESS in 3m 01s\n- tempest-full http://logs.openstack.org/56/652056/4/check/tempest-full/c47431f/ : SUCCESS in 2h 01m 14s\n- tempest-full-py3 http://logs.openstack.org/56/652056/4/check/tempest-full-py3/6e59099/ : SUCCESS in 1h 21m 51s\n- openstack-tox-cover http://logs.openstack.org/56/652056/4/check/openstack-tox-cover/c57fc6e/cover/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/4/check/openstack-tox-lower-constraints/7df6a3c/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/4/check/openstack-tox-pep8/6e8f489/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/4/check/openstack-tox-py27/45717fc/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/4/check/openstack-tox-py36/adf3f5c/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/4/check/openstack-tox-py37/4147257/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/56/652056/4/check/openstack-tox-docs/b47f2bc/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/4/check/build-openstack-releasenotes/a51accd/html/ : SUCCESS in 3m 31s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-rabbit/dc6c600/ : FAILURE in 5m 57s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-amqp1/6e7030f/ : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py27-func-kafka/f6f6076/ : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-rabbit/7b058db/ : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-amqp1/7d125fb/ : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-tox-py35-func-kafka/708b282/ : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-rabbit/14c7067/ : SUCCESS in 1h 41m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/14d4d9d/ : SUCCESS in 1h 41m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/c149960/ : SUCCESS in 2h 11m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-hybrid/1936bce/ : FAILURE in 20m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5bb76a3/ : FAILURE in 24m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm/e0ac5b6/ : SUCCESS in 1h 11m 32s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/4/check/oslo.messaging-src-grenade-dsvm-multinode/198b1dd/ : SUCCESS in 1h 11m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/575e921/ : FAILURE in 56m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/5c052d9/ : FAILURE in 58m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/d90a09a/ : FAILURE in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b64bffa6977731c022098d33e578cb896718cffa","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-16 10:56:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d700fc7911096a3b3ecea5e8972bdbb5ac707a11","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-17 14:43:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8b996cff936e074b86a56073df66a57d1e570abd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 16:53:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/56/652056/5/check/requirements-check/6f28267/ : SUCCESS in 3m 19s\n- tempest-full http://logs.openstack.org/56/652056/5/check/tempest-full/9442a8a/ : SUCCESS in 1h 41m 51s\n- tempest-full-py3 http://logs.openstack.org/56/652056/5/check/tempest-full-py3/40cccb0/ : SUCCESS in 1h 51m 58s\n- openstack-tox-cover http://logs.openstack.org/56/652056/5/check/openstack-tox-cover/d266c7e/cover/ : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/5/check/openstack-tox-lower-constraints/2b40347/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/5/check/openstack-tox-pep8/53834e7/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/5/check/openstack-tox-py27/e0ac248/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/5/check/openstack-tox-py36/5a9dedc/ : SUCCESS in 5m 21s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/5/check/openstack-tox-py37/114190f/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/56/652056/5/check/openstack-tox-docs/f9b7804/html/ : SUCCESS in 3m 18s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/5/check/build-openstack-releasenotes/cfeb90d/html/ : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py27-func-rabbit/c639bd4/ : SUCCESS in 5m 33s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py27-func-amqp1/bf61a5e/ : SUCCESS in 4m 44s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py27-func-kafka/6650df3/ : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py35-func-rabbit/66e16fa/ : SUCCESS in 5m 58s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py35-func-amqp1/eaf586c/ : SUCCESS in 4m 52s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/56/652056/5/check/oslo.messaging-tox-py35-func-kafka/0433af2/ : FAILURE in 8m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-dsvm-full-rabbit/1aeea18/ : SUCCESS in 1h 49m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/cdead5a/ : SUCCESS in 1h 34m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/9a0cb98/ : SUCCESS in 2h 02m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-dsvm-full-kafka-hybrid/83ceb01/ : FAILURE in 21m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-dsvm-full-kafka-centos-7/56b7805/ : FAILURE in 22m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-grenade-dsvm/5867e52/ : SUCCESS in 59m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/5/check/oslo.messaging-src-grenade-dsvm-multinode/e50f425/ : SUCCESS in 1h 09m 27s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/5/check/oslo.messaging-telemetry-dsvm-integration-rabbit/c461d22/ : FAILURE in 1h 06m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/5/check/oslo.messaging-telemetry-dsvm-integration-amqp1/be77540/ : FAILURE in 1h 13m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/5/check/oslo.messaging-telemetry-dsvm-integration-kafka/1f78c7b/ : FAILURE in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"45b849ddf6fa0c7f100d20e3dff28e30bd90374b","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-18 09:33:03.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"10e57ed49d36984e16649da313ab3cf0d487cc54","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-18 09:34:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c3efa076f8fec87177b8ca5512a5d43fd74b84aa","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-18 09:36:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"90d223eb1ed76ecb51ee6647329b658b07dfe7b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-18 09:39:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b09fd60e96e5beb17b8158600cf8bcfef68253fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-18 10:50:02.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nI\u0027m still not sure that we can remove exception translation, \nHervé. Could you bring this up on the mailing list, please?","accounts_in_message":[],"_revision_number":7},{"id":"8240a8cf66de8fac32e3f1ba74728976501426de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 11:58:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/652056/7/check/requirements-check/f4b9168/ : SUCCESS in 3m 57s\n- tempest-full http://logs.openstack.org/56/652056/7/check/tempest-full/768689d/ : SUCCESS in 1h 44m 15s\n- tempest-full-py3 http://logs.openstack.org/56/652056/7/check/tempest-full-py3/4b6b699/ : SUCCESS in 1h 43m 26s\n- openstack-tox-cover http://logs.openstack.org/56/652056/7/check/openstack-tox-cover/faee15f/cover/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/7/check/openstack-tox-lower-constraints/b7f34ca/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/7/check/openstack-tox-pep8/3d7c9ae/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/7/check/openstack-tox-py27/10fed17/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/7/check/openstack-tox-py36/74d6666/ : SUCCESS in 6m 46s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/7/check/openstack-tox-py37/b95e83b/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/56/652056/7/check/openstack-tox-docs/78402ff/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/7/check/build-openstack-releasenotes/d54d641/html/ : SUCCESS in 4m 51s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py27-func-rabbit/db92f32/ : FAILURE in 7m 53s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py27-func-amqp1/77183ce/ : FAILURE in 12m 31s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py27-func-kafka/32e8d8c/ : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py36-func-rabbit/f6e4a21/ : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py36-func-amqp1/c882be1/ : FAILURE in 12m 06s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/56/652056/7/check/oslo.messaging-tox-py36-func-kafka/e58c55c/ : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-dsvm-full-rabbit/9622754/ : SUCCESS in 1h 40m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/404a0a3/ : SUCCESS in 2h 10m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/da0896d/ : SUCCESS in 2h 04m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-dsvm-full-kafka-hybrid/71a1c2a/ : FAILURE in 18m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-dsvm-full-kafka-centos-7/41adb9e/ : FAILURE in 25m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-grenade-dsvm/a3a5b9c/ : SUCCESS in 1h 04m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/7/check/oslo.messaging-src-grenade-dsvm-multinode/c6cbe11/ : SUCCESS in 1h 09m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/7/check/oslo.messaging-telemetry-dsvm-integration-rabbit/98d2522/ : FAILURE in 1h 08m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/7/check/oslo.messaging-telemetry-dsvm-integration-amqp1/e892b6a/ : FAILURE in 1h 09m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/7/check/oslo.messaging-telemetry-dsvm-integration-kafka/6ea1dd5/ : FAILURE in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"de29ff961488785e4376d445d154b65d291d9785","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-18 13:28:39.000000000","message":"Patch Set 7:\n\n\u003e (4 comments)\n \u003e \n \u003e I\u0027m still not sure that we can remove exception translation,\n \u003e Hervé. Could you bring this up on the mailing list, please?\n\nYeah sure I\u0027ll ask soon on the ML about exception tranlations.","accounts_in_message":[],"_revision_number":7},{"id":"677f2e0a44bec1a97d535f2b6403e705ee81cf0f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-04-18 14:36:12.000000000","message":"Patch Set 7:\n\nRegarding exception translation, I think the rule of thumb followed by other projects is that if the exception message may be returned as part of an API response then they translate it, otherwise not. This doesn\u0027t appear likely to appear in an API call so I\u0027d say let\u0027s just drop the translation.","accounts_in_message":[],"_revision_number":7},{"id":"be456592292b09f45736cabaadbbe050dfc8e622","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-19 13:49:28.000000000","message":"Patch Set 7:\n\n\u003e Regarding exception translation, I think the rule of thumb followed\n \u003e by other projects is that if the exception message may be returned\n \u003e as part of an API response then they translate it, otherwise not.\n \u003e This doesn\u0027t appear likely to appear in an API call so I\u0027d say\n \u003e let\u0027s just drop the translation.\n\nOk thanks for info!","accounts_in_message":[],"_revision_number":7},{"id":"57c99c9db1b493fcbc67b9bf963d09952eda13c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-23 13:35:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"76014459933abd673508c9b4bc71356e726b6db6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 16:31:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/56/652056/8/check/requirements-check/16dbad3/ : SUCCESS in 4m 54s\n- tempest-full finger://ze09.openstack.org/8750f0c5b27f4803a4a6cfcc9af18189 : RETRY_LIMIT in 21m 36s\n- tempest-full-py3 finger://ze05.openstack.org/d038e7b5da9b4602b2a333720c4f5997 : RETRY_LIMIT in 28m 49s\n- openstack-tox-cover http://logs.openstack.org/56/652056/8/check/openstack-tox-cover/d02d7a8/cover/ : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/8/check/openstack-tox-lower-constraints/4687500/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/8/check/openstack-tox-pep8/f85bbf0/ : FAILURE in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/8/check/openstack-tox-py27/5c4a413/ : SUCCESS in 7m 02s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/8/check/openstack-tox-py36/e0045ba/ : SUCCESS in 8m 36s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/8/check/openstack-tox-py37/b6b4619/ : SUCCESS in 8m 30s\n- openstack-tox-docs http://logs.openstack.org/56/652056/8/check/openstack-tox-docs/a0b2be1/html/ : SUCCESS in 6m 02s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/8/check/build-openstack-releasenotes/14e13bd/html/ : SUCCESS in 5m 01s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py27-func-rabbit/bdad97e/ : SUCCESS in 7m 27s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py27-func-amqp1/73c5a94/ : FAILURE in 13m 28s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py27-func-kafka/bb8684a/ : FAILURE in 5m 39s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py36-func-rabbit/39e098b/ : SUCCESS in 6m 39s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py36-func-amqp1/2631db9/ : FAILURE in 12m 46s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/56/652056/8/check/oslo.messaging-tox-py36-func-kafka/dc67d24/ : FAILURE in 6m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit finger://ze07.openstack.org/f05db6aedac54d73a841350fe02f4c86 : RETRY_LIMIT in 23m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid finger://ze07.openstack.org/29cd70bae42d434dbe43a4741b650d53 : RETRY_LIMIT in 27m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 finger://ze12.openstack.org/b7f2958fdc044d9c996f50d9417bb573 : RETRY_LIMIT in 18m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/8/check/oslo.messaging-src-dsvm-full-kafka-hybrid/7a0007c/ : FAILURE in 18m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/8/check/oslo.messaging-src-dsvm-full-kafka-centos-7/d8abdda/ : FAILURE in 26m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm finger://ze01.openstack.org/5048213359cd4d05afbece1c82c6dadf : RETRY_LIMIT in 30m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode finger://ze11.openstack.org/75c02c798082420994efb14806991803 : RETRY_LIMIT in 14m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit finger://ze07.openstack.org/c17b53b4212147df8a9dff69b5c00b2d : RETRY_LIMIT in 14m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 finger://ze12.openstack.org/7b4bce710d1746d7af9e1fa3265cbeb8 : RETRY_LIMIT in 24m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/8/check/oslo.messaging-telemetry-dsvm-integration-kafka/95eafb4/ : FAILURE in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9622f608783e70d3ecd9af69917bf5bf1499f76a","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-23 19:50:26.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"5e3ed690b1c2c164b943f39e22286291c8e5bda9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-24 09:55:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b9980dd24560cefdae4454c7bd8e0e5b43cc675c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:13:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/56/652056/9/check/requirements-check/311c41d/ : SUCCESS in 3m 20s\n- tempest-full http://logs.openstack.org/56/652056/9/check/tempest-full/51300ed/ : SUCCESS in 1h 38m 43s\n- tempest-full-py3 http://logs.openstack.org/56/652056/9/check/tempest-full-py3/a992d50/ : SUCCESS in 1h 32m 09s\n- openstack-tox-cover http://logs.openstack.org/56/652056/9/check/openstack-tox-cover/94cfeed/cover/ : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/9/check/openstack-tox-lower-constraints/38d1f25/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/9/check/openstack-tox-pep8/9a30030/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/9/check/openstack-tox-py27/48b7105/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/9/check/openstack-tox-py36/1d06ea2/ : SUCCESS in 5m 10s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/9/check/openstack-tox-py37/fa4770c/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/56/652056/9/check/openstack-tox-docs/32ca9e8/html/ : SUCCESS in 3m 12s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/9/check/build-openstack-releasenotes/3704e85/html/ : SUCCESS in 3m 13s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py27-func-rabbit/fc18284/ : SUCCESS in 6m 06s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py27-func-amqp1/794dc27/ : FAILURE in 11m 50s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py27-func-kafka/3ba3fca/ : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py36-func-rabbit/1d46b5c/ : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py36-func-amqp1/77f1f8e/ : FAILURE in 12m 15s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/56/652056/9/check/oslo.messaging-tox-py36-func-kafka/93aa0c5/ : FAILURE in 4m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-dsvm-full-rabbit/ed2129b/ : SUCCESS in 1h 30m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/af38910/ : SUCCESS in 1h 32m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/5eb0dd8/ : SUCCESS in 2h 02m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-dsvm-full-kafka-hybrid/5e29d12/ : SUCCESS in 1h 34m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-dsvm-full-kafka-centos-7/baeff85/ : SUCCESS in 2h 08m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-grenade-dsvm/d70e9da/ : SUCCESS in 1h 13m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/56/652056/9/check/oslo.messaging-src-grenade-dsvm-multinode/e448a0f/ : SUCCESS in 1h 14m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/56/652056/9/check/oslo.messaging-telemetry-dsvm-integration-rabbit/ad68865/ : FAILURE in 56m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/56/652056/9/check/oslo.messaging-telemetry-dsvm-integration-amqp1/2ffd9eb/ : FAILURE in 58m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/56/652056/9/check/oslo.messaging-telemetry-dsvm-integration-kafka/5d10b75/ : FAILURE in 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9b64101a83925d73a36924909ea4f054008cc345","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-24 13:17:20.000000000","message":"Patch Set 9: Code-Review+2\n\nIf bnemec is happy with this then so am I. Passing CI now and I can\u0027t see anything else off","accounts_in_message":[],"_revision_number":9},{"id":"125c8f52165494c96c74930dc0bbf8fccc2fe8e9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-04 20:23:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nYep, this works for me. Thanks!","accounts_in_message":[],"_revision_number":9},{"id":"2504052f77c2603b1b94b008a0790e6e4203bc2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 20:23:13.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1efc535496372e7e0c26114f57d09a3e665291a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:31:13.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/56/652056/9/gate/requirements-check/4ebac5b/ : SUCCESS in 4m 28s\n- tempest-full http://logs.openstack.org/56/652056/9/gate/tempest-full/34b8c4d/ : SUCCESS in 2h 05m 40s\n- tempest-full-py3 http://logs.openstack.org/56/652056/9/gate/tempest-full-py3/0bba648/ : SUCCESS in 1h 26m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/652056/9/gate/openstack-tox-lower-constraints/d4a8d3b/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/56/652056/9/gate/openstack-tox-pep8/e6840bb/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/56/652056/9/gate/openstack-tox-py27/67fb9a0/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/56/652056/9/gate/openstack-tox-py36/975f686/ : SUCCESS in 4m 57s\n- openstack-tox-py37 http://logs.openstack.org/56/652056/9/gate/openstack-tox-py37/91e0559/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/56/652056/9/gate/openstack-tox-docs/c81ac2d/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/56/652056/9/gate/build-openstack-releasenotes/517f0ab/html/ : SUCCESS in 3m 29s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/56/652056/9/gate/oslo.messaging-tox-py27-func-rabbit/c527b83/ : SUCCESS in 5m 33s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/56/652056/9/gate/oslo.messaging-src-dsvm-full-rabbit/3e4ff79/ : SUCCESS in 2h 01m 47s","accounts_in_message":[],"_revision_number":9},{"id":"875dacae07fc8559eff79b1d019b7027c5c4f2ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"4f385720d7d43aadfaff7c6f41b0534312b23ba8","revisions":{"b0e595ab56f4d4219635a336a285f811e33d9209":{"kind":"REWORK","_number":1,"created":"2019-04-12 12:35:48.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/1"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-12 12:33:58.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b0e595ab56f4d4219635a336a285f811e33d9209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b0e595ab56f4d4219635a336a285f811e33d9209"}]},"branch":"refs/heads/master"},"58de883e2646a6d055639f1d4a839659cb8d136a":{"kind":"REWORK","_number":2,"created":"2019-04-12 12:46:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/2"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-12 12:46:01.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/58de883e2646a6d055639f1d4a839659cb8d136a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/58de883e2646a6d055639f1d4a839659cb8d136a"}]},"branch":"refs/heads/master"},"1f6cd6bbd0307f8b062ea3f1a613224b55649e57":{"kind":"REWORK","_number":3,"created":"2019-04-12 13:56:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/3"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-12 13:53:55.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f6cd6bbd0307f8b062ea3f1a613224b55649e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f6cd6bbd0307f8b062ea3f1a613224b55649e57"}]},"branch":"refs/heads/master"},"72636d5ddc0517d4c0c73e80157f4b17a5bfa1a2":{"kind":"REWORK","_number":4,"created":"2019-04-15 10:13:46.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/4"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-15 10:13:22.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/72636d5ddc0517d4c0c73e80157f4b17a5bfa1a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/72636d5ddc0517d4c0c73e80157f4b17a5bfa1a2"}]},"branch":"refs/heads/master"},"01ffe37f40ed04d45a683fe1708e7b71dd4fdc19":{"kind":"REWORK","_number":5,"created":"2019-04-17 14:43:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/5"}}},"commit":{"parents":[{"commit":"e99af8ca576b7ff56eadbe6351216c2d0d2b93c7","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e99af8ca576b7ff56eadbe6351216c2d0d2b93c7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-17 14:43:35.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01ffe37f40ed04d45a683fe1708e7b71dd4fdc19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01ffe37f40ed04d45a683fe1708e7b71dd4fdc19"}]},"branch":"refs/heads/master"},"43cee045e789b9aa963d57b94c739a6df728eee9":{"kind":"REWORK","_number":6,"created":"2019-04-18 09:34:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/6"}}},"commit":{"parents":[{"commit":"d4873f983bf79665d7c2cfe02eecdbe972626cc7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d4873f983bf79665d7c2cfe02eecdbe972626cc7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-18 09:34:15.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43cee045e789b9aa963d57b94c739a6df728eee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43cee045e789b9aa963d57b94c739a6df728eee9"}]},"branch":"refs/heads/master"},"bba51a2858deb1805faa722545f2ac04431eb37a":{"kind":"REWORK","_number":7,"created":"2019-04-18 09:39:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/7"}}},"commit":{"parents":[{"commit":"d4873f983bf79665d7c2cfe02eecdbe972626cc7","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d4873f983bf79665d7c2cfe02eecdbe972626cc7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-18 09:39:24.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bba51a2858deb1805faa722545f2ac04431eb37a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bba51a2858deb1805faa722545f2ac04431eb37a"}]},"branch":"refs/heads/master"},"af67413d580d50b8898153f90982730bb59160d0":{"kind":"REWORK","_number":8,"created":"2019-04-23 13:35:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/8"}}},"commit":{"parents":[{"commit":"433d34ed4149ad2dcbb860b07754dea25cb43407","subject":"Merge \"Retry to declare a queue after internal error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/433d34ed4149ad2dcbb860b07754dea25cb43407"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-23 13:35:31.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af67413d580d50b8898153f90982730bb59160d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af67413d580d50b8898153f90982730bb59160d0"}]},"branch":"refs/heads/master"},"4f385720d7d43aadfaff7c6f41b0534312b23ba8":{"kind":"REWORK","_number":9,"created":"2019-04-24 09:55:51.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/56/652056/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/56/652056/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/56/652056/9"}}},"commit":{"parents":[{"commit":"433d34ed4149ad2dcbb860b07754dea25cb43407","subject":"Merge \"Retry to declare a queue after internal error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/433d34ed4149ad2dcbb860b07754dea25cb43407"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-11 08:47:24.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-04-24 09:55:38.000000000","tz":120},"subject":"Remove log translation and i18n","message":"Remove log translation and i18n\n\nLog messages are no longer being translated. This removes all use of\nthe _LE, _LI, and _LW translation markers to simplify logging and to\navoid confusion with new contributions.\n\nChange-Id: I9ddb6595fc52e46ed8844e39d2fa71029c90b65c\nCloses-Bug: #1674567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f385720d7d43aadfaff7c6f41b0534312b23ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f385720d7d43aadfaff7c6f41b0534312b23ba8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
