)]}'
{"id":"openstack%2Foslo.messaging~654817","triplet_id":"openstack%2Foslo.messaging~stable%2Fpike~I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","project":"openstack/oslo.messaging","branch":"stable/pike","topic":"bug/1822778-stable/pike","hashtags":[],"change_id":"I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","subject":"Retry to declare a queue after internal error","status":"ABANDONED","created":"2019-04-23 07:49:37.000000000","updated":"2019-12-18 11:05:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dbe8432a2c676a3af2637a8c8b18500eab10fea","_number":654817,"virtual_id_number":654817,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-04-23 12:22:51.000000000","_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"date":"2019-06-26 14:38:51.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"date":"2019-05-03 19:51:42.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2019-04-23 09:49:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 09:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-03 19:51:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-26 14:38:51.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"53638e06ee4d7e733d5d644fcb12e3151827a154","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:49:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"86c29c18d6ff413039b8828414c6c0b71ec510d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:49:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/17/654817/1/check/openstack-tox-pep8/1b52c83/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/17/654817/1/check/openstack-tox-py27/f3d6f36/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/17/654817/1/check/openstack-tox-py35/b92de55/ : FAILURE in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/17/654817/1/check/openstack-tox-docs/51ed1e3/html/ : SUCCESS in 3m 31s\n- tempest-full http://logs.openstack.org/17/654817/1/check/tempest-full/db26f51/ : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e66bce4691b99f671150e1302ca77c915164ab23","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 12:22:51.000000000","message":"Patch Set 1:\n\ntests fail on /zmq_common.py:\nhttp://logs.openstack.org/17/654817/1/check/openstack-tox-py35/b92de55/testr_results.html.gz\nnot related with the fix","accounts_in_message":[],"_revision_number":1},{"id":"7763e9f44bf60949a253aa0fd632a5f227be2307","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 19:51:42.000000000","message":"Patch Set 1:\n\nI think pike is essentially EOL now so we can probably drop this","accounts_in_message":[],"_revision_number":1},{"id":"be5c8abc7f1f3bf819f4bf3684a57c6e4d96ca88","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-29 17:31:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b00472de9ca13f04b336fe3e2daf79a58d29ed04","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-29 17:38:41.000000000","message":"Patch Set 1:\n\n\u003e I think pike is essentially EOL now so we can probably drop this\n\nIs it EOL? Isn\u0027t oslo.messaging stable/pike open for bugfixes (Extended Maintenance)?","accounts_in_message":[],"_revision_number":1},{"id":"4fd8718479312d509c70d0bb175baeec0eb0148f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:10:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/17/654817/1/check/openstack-tox-pep8/f1e21e9/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/17/654817/1/check/openstack-tox-py27/9f8b635/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/17/654817/1/check/openstack-tox-py35/de44d7e/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/17/654817/1/check/openstack-tox-docs/3f44de8/ : POST_FAILURE in 2m 34s\n- tempest-full http://logs.openstack.org/17/654817/1/check/tempest-full/b78159e/ : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":1},{"id":"23f7e8a00a859646864ebf5c0571f34dad773278","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-06-26 14:38:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2be99f652656ee4432b821dc2d5a07685e808975","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 16:13:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/17/654817/1/check/openstack-tox-pep8/238cee9/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/17/654817/1/check/openstack-tox-py27/f3b6dc9/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/17/654817/1/check/openstack-tox-py35/7d205c0/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/17/654817/1/check/openstack-tox-docs/1be9e09/ : POST_FAILURE in 2m 34s\n- tempest-full http://logs.openstack.org/17/654817/1/check/tempest-full/a3a5c41/ : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"75f9a4a9b23769ecb62528fb7234f1b77010ebc0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 11:05:18.000000000","message":"Abandoned\n\nPike is in extended maintenance now so this won\u0027t ever be released. I\u0027m going to close this as it doesn\u0027t seem worth the effort to merge now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0297c5f0ccfa411aded1a1cb3559de90aa276dd3","revisions":{"0297c5f0ccfa411aded1a1cb3559de90aa276dd3":{"kind":"REWORK","_number":1,"created":"2019-04-23 07:49:37.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/17/654817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/654817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/654817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/654817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/654817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/654817/1"}}},"commit":{"parents":[{"commit":"42275b21b405bbe76847eb242b1c5166c1c4a0a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42275b21b405bbe76847eb242b1c5166c1c4a0a6"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","date":"2019-04-23 07:49:37.000000000","tz":0},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: #1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n(cherry picked from commit 4d2787227b00b973973554f7387e621d2664c0d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0297c5f0ccfa411aded1a1cb3559de90aa276dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0297c5f0ccfa411aded1a1cb3559de90aa276dd3"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
