)]}'
{"id":"openstack%2Foslo.messaging~654818","triplet_id":"openstack%2Foslo.messaging~stable%2Fqueens~I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","project":"openstack/oslo.messaging","branch":"stable/queens","topic":"bug/1822778-stable/queens","hashtags":[],"change_id":"I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","subject":"Retry to declare a queue after internal error","status":"MERGED","created":"2019-04-23 07:50:24.000000000","updated":"2019-05-04 15:20:07.000000000","submitted":"2019-05-04 15:20:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654818-1556983207066-22df3888","meta_rev_id":"2c446380fc980d696f68bc10a8fe36809535c224","_number":654818,"virtual_id_number":654818,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-04 15:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-04-24 13:08:24.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2019-05-03 19:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-04-23 16:20:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-03 19:50:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-04 12:15:16.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 16:20:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-04-24 13:08:24.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-03 19:50:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-04 12:15:16.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-05-04 15:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"854d10ddd544c7fc59d8dfae594eac82736270ee","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:50:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0fd320d7527c852f636e9b06e1be23875de82ae4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:11:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/654818/1/check/openstack-tox-pep8/65bc93d/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/18/654818/1/check/openstack-tox-py27/4dc5419/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/18/654818/1/check/openstack-tox-py35/878d6dc/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/18/654818/1/check/openstack-tox-docs/5cf1a76/html/ : SUCCESS in 3m 12s\n- tempest-full http://logs.openstack.org/18/654818/1/check/tempest-full/3d65f1a/ : SUCCESS in 1h 25m 51s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-amqp1/0d6fc2d/ : SUCCESS in 4m 21s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-kafka/835cb25/ : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-pika/e96b6c9/ : SUCCESS in 6m 21s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-rabbit/c927b04/ : SUCCESS in 4m 51s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq-proxy/bf4280b/ : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq-pubsub/99ec9d3/ : SUCCESS in 4m 46s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq/7cb38f5/ : TIMED_OUT in 41m 28s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-amqp1/1b8ccb9/ : FAILURE in 4m 26s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-rabbit/bcdb842/ : SUCCESS in 5m 07s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-zmq/afb9b77/ : SUCCESS in 4m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-rabbit-default/3a356b0/ : SUCCESS in 1h 13m 57s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/ddadb16/ : FAILURE in 29m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/5ae7853/ : SUCCESS in 1h 39m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/8ea05b0/ : FAILURE in 28m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-kafka-default/e5d929d/ : FAILURE in 15m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-pika-default/e4fefd5/ : SUCCESS in 1h 15m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-zmq-default/67d2556/ : FAILURE in 18m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-grenade-dsvm/b07adae/ : SUCCESS in 1h 00m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-grenade-dsvm-multinode/5c515ff/ : SUCCESS in 55m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/a920474/ : SUCCESS in 41m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/8f86dba/ : FAILURE in 38m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/03fa9c9/ : FAILURE in 18m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-pika/789a3c9/ : FAILURE in 45m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-zmq/909f698/ : FAILURE in 18m 51s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/9f2df92/ : SUCCESS in 1h 17m 02s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/78a736c/ : FAILURE in 31m 37s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/7b93599/ : FAILURE in 18m 33s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/565e917/ : SUCCESS in 1h 15m 22s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/911b3ca/ : FAILURE in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1579606e821f1693b7fa9ca89e6e04ee261500f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-04-23 16:20:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"577f09e904c0a0b189b945637fb71593f76f3960","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-24 13:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79d9a9cf6fda9af99126eabfb3d4e2eb0b05c0f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 19:50:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"5feaead10874a637f71bd08d1be7a68a8825b243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:51:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0aa89736ce569a00784aae3b55cf10dc9e29964e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:36:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/18/654818/1/gate/openstack-tox-pep8/8304c2a/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/18/654818/1/gate/openstack-tox-py27/e853ac0/ : SUCCESS in 5m 25s\n- openstack-tox-py35 http://logs.openstack.org/18/654818/1/gate/openstack-tox-py35/442d741/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/18/654818/1/gate/openstack-tox-docs/4ef6931/html/ : SUCCESS in 3m 08s\n- tempest-full http://logs.openstack.org/18/654818/1/gate/tempest-full/6c89473/ : SUCCESS in 1h 23m 55s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tox-py27-func-rabbit/256eea7/ : SUCCESS in 4m 21s\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tox-py27-func-pika/44cc5ac/ : SUCCESS in 4m 38s\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/18/654818/1/gate/oslo.messaging-src-dsvm-full-rabbit-default/71183dd/ : SUCCESS in 1h 17m 39s\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/0d146bc/ : FAILURE in 1h 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ffd91f61e73b133054a8e2cab5d5fbb34803f6be","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-05-04 12:15:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64450f00220f4105eb4d717aec6fb6a138b03ee4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 13:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/654818/1/check/openstack-tox-pep8/e5e3315/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/18/654818/1/check/openstack-tox-py27/327f6d8/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/18/654818/1/check/openstack-tox-py35/5be66b4/ : SUCCESS in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/18/654818/1/check/openstack-tox-docs/29e8298/html/ : SUCCESS in 3m 08s\n- tempest-full http://logs.openstack.org/18/654818/1/check/tempest-full/fa710dc/ : SUCCESS in 1h 26m 13s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-amqp1/c9cdff9/ : SUCCESS in 4m 26s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-kafka/251861b/ : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-pika/0793753/ : SUCCESS in 4m 24s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-rabbit/60244d1/ : SUCCESS in 5m 39s\n- oslo.messaging-tox-py27-func-zmq-proxy http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq-proxy/f03b762/ : SUCCESS in 4m 16s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq-pubsub/eedd7ff/ : FAILURE in 4m 56s (non-voting)\n- oslo.messaging-tox-py27-func-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py27-func-zmq/30096d6/ : SUCCESS in 5m 17s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-amqp1/4e6036e/ : SUCCESS in 4m 56s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-rabbit/1b4c23e/ : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-tox-py35-func-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-tox-py35-func-zmq/1c7f5ea/ : SUCCESS in 5m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-rabbit-default/db09612/ : SUCCESS in 1h 11m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/157d4c6/ : FAILURE in 28m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-amqp1-dual-centos-7/82d3300/ : SUCCESS in 1h 29m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-kafka-default-centos-7/14aecc6/ : FAILURE in 21m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-kafka-default/9478096/ : FAILURE in 17m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-pika-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-pika-default/739ab41/ : SUCCESS in 1h 08m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-zmq-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-dsvm-full-zmq-default/6e757d0/ : FAILURE in 19m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-grenade-dsvm/40cd1c8/ : SUCCESS in 59m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/18/654818/1/check/oslo.messaging-src-grenade-dsvm-multinode/407fa19/ : SUCCESS in 1h 02m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/590d767/ : FAILURE in 36m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/6e512c1/ : FAILURE in 26m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/e7e3216/ : FAILURE in 21m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-pika http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-pika/9a25505/ : FAILURE in 44m 37s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-zmq http://logs.openstack.org/18/654818/1/check/oslo.messaging-telemetry-dsvm-integration-zmq/6c89f1f/ : FAILURE in 19m 29s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/b06018e/ : SUCCESS in 1h 20m 01s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid/4a70c01/ : FAILURE in 28m 35s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-kafka-default/ce95eab/ : FAILURE in 12m 18s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-pika-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-pika-default/3ddd180/ : SUCCESS in 1h 15m 16s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-zmq-default http://logs.openstack.org/18/654818/1/check/oslo.messaging-tempest-neutron-dsvm-src-zmq-default/6e188f1/ : FAILURE in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b071cee0bca9c0624ef80d1608a8b71aaf8e2968","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 13:54:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13e1a500d960a852e0c46cb36026afa841df59f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 15:20:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/18/654818/1/gate/openstack-tox-pep8/9c34c1b/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/18/654818/1/gate/openstack-tox-py27/d8b6765/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/18/654818/1/gate/openstack-tox-py35/4fce2f4/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/18/654818/1/gate/openstack-tox-docs/c551e6e/html/ : SUCCESS in 3m 25s\n- tempest-full http://logs.openstack.org/18/654818/1/gate/tempest-full/c6e41f2/ : SUCCESS in 1h 24m 15s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tox-py27-func-rabbit/15836d4/ : SUCCESS in 4m 03s\n- oslo.messaging-tox-py27-func-pika http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tox-py27-func-pika/300ee7a/ : SUCCESS in 4m 10s\n- oslo.messaging-src-dsvm-full-rabbit-default http://logs.openstack.org/18/654818/1/gate/oslo.messaging-src-dsvm-full-rabbit-default/bff19fa/ : SUCCESS in 1h 19m 23s\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default http://logs.openstack.org/18/654818/1/gate/oslo.messaging-tempest-neutron-dsvm-src-rabbit-default/8a5e4e3/ : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2d60b5a7a13b95af3add935844a6133cdc3c09e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 15:20:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea1bb5fad84079d07d8ad48d2c2a71aa29cdda4","revisions":{"6ea1bb5fad84079d07d8ad48d2c2a71aa29cdda4":{"kind":"REWORK","_number":1,"created":"2019-04-23 07:50:24.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/18/654818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/654818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/654818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/654818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/654818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/654818/1"}}},"commit":{"parents":[{"commit":"50b4451503122fc100fa1d8714d052267a0d5650","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/50b4451503122fc100fa1d8714d052267a0d5650"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","date":"2019-04-23 07:50:24.000000000","tz":0},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: #1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n(cherry picked from commit 4d2787227b00b973973554f7387e621d2664c0d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6ea1bb5fad84079d07d8ad48d2c2a71aa29cdda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6ea1bb5fad84079d07d8ad48d2c2a71aa29cdda4"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
