)]}'
{"id":"openstack%2Foslo.messaging~654819","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bug/1822778-stable/rocky","hashtags":[],"change_id":"I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","subject":"Retry to declare a queue after internal error","status":"MERGED","created":"2019-04-23 07:50:41.000000000","updated":"2019-05-03 21:36:15.000000000","submitted":"2019-05-03 21:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654819-1556919375083-13d10e42","meta_rev_id":"9e57ffe263cb2064c41ad85763aeafa278ce2d15","_number":654819,"virtual_id_number":654819,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-03 21:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-04-24 13:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2019-05-03 19:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-03 19:50:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 13:06:42.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-03 19:50:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-03 21:36:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"613a8f4da8a18ca13c6f17b12495460e3f46e892","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:50:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ae83028d5f692fe5a2be63d45714cacf3c0adf28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/19/654819/1/check/openstack-tox-pep8/14b3c2e/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/19/654819/1/check/openstack-tox-py27/e9bcc9d/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/19/654819/1/check/openstack-tox-py35/1d5fdd7/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/19/654819/1/check/openstack-tox-docs/9c2c73b/html/ : SUCCESS in 3m 41s\n- tempest-full http://logs.openstack.org/19/654819/1/check/tempest-full/92c128a/ : SUCCESS in 1h 47m 08s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py27-func-rabbit/1e6249f/ : SUCCESS in 4m 46s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py27-func-amqp1/02bf938/ : SUCCESS in 4m 37s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py27-func-kafka/4dd6089/ : FAILURE in 4m 41s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py35-func-rabbit/2c60976/ : SUCCESS in 5m 02s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py35-func-amqp1/d1b8706/ : SUCCESS in 5m 03s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/19/654819/1/check/oslo.messaging-tox-py35-func-kafka/980cc72/ : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-dsvm-full-rabbit/9829717/ : SUCCESS in 1h 34m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/3045ff5/ : FAILURE in 34m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/cb5c8af/ : FAILURE in 1h 33m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/d5dc90e/ : FAILURE in 30m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/a507a2b/ : SUCCESS in 1h 38m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-grenade-dsvm/919206f/ : SUCCESS in 1h 09m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/19/654819/1/check/oslo.messaging-src-grenade-dsvm-multinode/2ce43af/ : SUCCESS in 1h 05m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/19/654819/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/7879e97/ : SUCCESS in 54m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/19/654819/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/6b5ca08/ : FAILURE in 46m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/19/654819/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/ff3d5d4/ : FAILURE in 37m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/19/654819/1/check/openstack-tox-lower-constraints/7ded312/ : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"531f540a95cb82922da58a0dbf2a1bdec40db5bd","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-24 13:06:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a00e21c54244723d6f9a34cf55d4a13a3369de1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 19:50:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"27e728d348fc46fcc5704b1e503e1edf76680870","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83f1c8bf9c31c54903043d383f9282803759eae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:36:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/19/654819/1/gate/openstack-tox-pep8/86426b9/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/19/654819/1/gate/openstack-tox-py27/0ca8e81/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/19/654819/1/gate/openstack-tox-py35/9338829/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/19/654819/1/gate/openstack-tox-docs/ae1edbb/html/ : SUCCESS in 3m 35s\n- tempest-full http://logs.openstack.org/19/654819/1/gate/tempest-full/0f20ad9/ : SUCCESS in 1h 40m 17s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/19/654819/1/gate/oslo.messaging-tox-py27-func-rabbit/942f5d9/ : SUCCESS in 5m 03s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/19/654819/1/gate/oslo.messaging-src-dsvm-full-rabbit/e34f161/ : SUCCESS in 1h 20m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/654819/1/gate/openstack-tox-lower-constraints/2b765e1/ : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f063bf741a6a915b4d60cdcf61c6db9305c83c27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 21:36:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d75eba02c74d309cd769a8257d2d20f735c60ad1","revisions":{"d75eba02c74d309cd769a8257d2d20f735c60ad1":{"kind":"REWORK","_number":1,"created":"2019-04-23 07:50:41.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/19/654819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/19/654819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/654819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/654819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/19/654819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/19/654819/1"}}},"commit":{"parents":[{"commit":"7c00cc2c2cefd938413276c05188e01e63ed46a7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c00cc2c2cefd938413276c05188e01e63ed46a7"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","date":"2019-04-23 07:50:41.000000000","tz":0},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: #1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n(cherry picked from commit 4d2787227b00b973973554f7387e621d2664c0d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d75eba02c74d309cd769a8257d2d20f735c60ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d75eba02c74d309cd769a8257d2d20f735c60ad1"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
