)]}'
{"id":"openstack%2Foslo.messaging~654820","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"bug/1822778-stable/stein","hashtags":[],"change_id":"I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a","subject":"Retry to declare a queue after internal error","status":"MERGED","created":"2019-04-23 07:51:00.000000000","updated":"2019-05-29 21:22:30.000000000","submitted":"2019-05-29 21:22:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654820-1559164950920-e2e73a0f","meta_rev_id":"95e6fcf3426b357fd11807681fbfaa53f355642e","_number":654820,"virtual_id_number":654820,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2019-05-29 17:28:46.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-05-29 21:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-04-24 13:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-05-03 19:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-05-03 19:51:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 13:07:21.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-03 19:51:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-29 17:28:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-05-29 21:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e80dc321b9ff27e6dd73f580beae144604106db","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-04-23 07:51:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"26e53e5dfa9e2cadb17652bcc321b8c6bfee3dc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 11:15:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/20/654820/1/check/tempest-full/1675867/ : SUCCESS in 1h 45m 34s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/check/tempest-full-py3/eafdd67/ : SUCCESS in 1h 29m 46s\n- openstack-tox-cover http://logs.openstack.org/20/654820/1/check/openstack-tox-cover/e677372/cover/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/check/openstack-tox-lower-constraints/34fee75/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/check/openstack-tox-pep8/32eb08e/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/check/openstack-tox-py27/daf6f4e/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/check/openstack-tox-py36/078aa22/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/check/openstack-tox-py37/10cc3cf/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/check/openstack-tox-docs/24231eb/html/ : SUCCESS in 4m 00s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-rabbit/286d6d0/ : SUCCESS in 5m 20s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-amqp1/7eba548/ : SUCCESS in 4m 42s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-kafka/6ba3cc3/ : FAILURE in 4m 44s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-rabbit/339e379/ : SUCCESS in 5m 44s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-amqp1/b57c7f5/ : SUCCESS in 4m 35s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-kafka/082bd86/ : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-rabbit/c174293/ : SUCCESS in 1h 33m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/87a06d2/ : SUCCESS in 1h 34m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/1474430/ : SUCCESS in 1h 51m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/771560a/ : FAILURE in 19m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5a047c9/ : FAILURE in 22m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm/71967da/ : SUCCESS in 55m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm-multinode/909f13d/ : SUCCESS in 1h 11m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/6c2bdfd/ : FAILURE in 57m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/f8dd891/ : FAILURE in 57m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/8216a56/ : FAILURE in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70d1bed72cc1cd70b24c7ac409b0650f56528ed6","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-24 13:07:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa0a1e95b8bfdfa757d03aa3fe6e991f10ab5c6f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-03 19:51:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"b1523cd557c5696efbf679b6700d2520faf5516c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:51:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bcd7ba7be2e1f3253517661a99ef2fd7ece8245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:07:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/20/654820/1/gate/tempest-full/1781125/ : SUCCESS in 1h 48m 26s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/gate/tempest-full-py3/4aee4d3/ : SUCCESS in 1h 27m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/gate/openstack-tox-lower-constraints/b987aa0/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/gate/openstack-tox-pep8/c1b64fe/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py27/5a1a902/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py36/a6ba3a5/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py37/862737d/ : SUCCESS in 6m 55s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/gate/openstack-tox-docs/8d180ab/html/ : SUCCESS in 3m 53s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-tox-py27-func-rabbit/d90a177/ : FAILURE in 6m 27s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-src-dsvm-full-rabbit/f618072/ : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"aab9c7d23248f65007a1a8fb032b5ca089d36da5","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-05 18:09:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b8c85d85cc8a6ae8bb1e873ae90825c25070654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 20:12:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/20/654820/1/check/tempest-full/06ddec4/ : SUCCESS in 1h 44m 53s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/check/tempest-full-py3/a925b90/ : SUCCESS in 1h 16m 38s\n- openstack-tox-cover http://logs.openstack.org/20/654820/1/check/openstack-tox-cover/f61e1f4/cover/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/check/openstack-tox-lower-constraints/39e097c/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/check/openstack-tox-pep8/b61fd1d/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/check/openstack-tox-py27/19d7dc4/ : SUCCESS in 6m 07s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/check/openstack-tox-py36/3d0636e/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/check/openstack-tox-py37/fbbdf66/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/check/openstack-tox-docs/80ed6db/html/ : SUCCESS in 3m 23s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-rabbit/5aabba6/ : SUCCESS in 4m 57s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-amqp1/220099a/ : SUCCESS in 4m 06s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-kafka/4d56269/ : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-rabbit/0e81269/ : SUCCESS in 4m 21s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-amqp1/7ca62dd/ : SUCCESS in 4m 19s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-kafka/0396b7b/ : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-rabbit/bdfc27f/ : SUCCESS in 1h 27m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/2e4c519/ : SUCCESS in 1h 27m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/1ca0aa4/ : SUCCESS in 1h 50m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/23eadfc/ : FAILURE in 19m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/4218400/ : FAILURE in 18m 43s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm/0e81986/ : SUCCESS in 54m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm-multinode/6f70e66/ : SUCCESS in 1h 06m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/7ed4261/ : FAILURE in 36m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/540ea21/ : FAILURE in 39m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/801387c/ : FAILURE in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54de16b3bb1a9fdd041d1993bb0617f88b23c14d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 20:12:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4020885bbca76c26bbf0879369cda94e4852c607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 21:41:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/20/654820/1/gate/tempest-full/196fc35/ : FAILURE in 27m 31s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/gate/tempest-full-py3/14e31b7/ : SUCCESS in 1h 28m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/gate/openstack-tox-lower-constraints/39349fe/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/gate/openstack-tox-pep8/09d9e8e/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py27/93bb389/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py36/b5f0ed4/ : SUCCESS in 4m 41s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py37/9f94d1b/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/gate/openstack-tox-docs/240c6d1/html/ : SUCCESS in 3m 10s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-tox-py27-func-rabbit/3745e17/ : SUCCESS in 5m 12s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-src-dsvm-full-rabbit/cc72577/ : FAILURE in 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9934b9e84ddc1bf0705798e66b9082040a059b12","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-29 15:04:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"671b9e8070d7e5e340f0dcb84e848785bdb7351a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-29 15:07:17.000000000","message":"Patch Set 1:\n\nThis patch was backported and merged to Rocky and Queens, so let\u0027s get this merged in Stein, too, to not to break stable rules.","accounts_in_message":[],"_revision_number":1},{"id":"42c7045899bc76b1fca6f5139319f8e736952e35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 17:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/20/654820/1/check/tempest-full/80759f1/ : FAILURE in 1h 48m 06s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/check/tempest-full-py3/39e3c9b/ : SUCCESS in 1h 22m 16s\n- openstack-tox-cover http://logs.openstack.org/20/654820/1/check/openstack-tox-cover/34c5de1/cover/ : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/check/openstack-tox-lower-constraints/01d8ec9/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/check/openstack-tox-pep8/d69d88f/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/check/openstack-tox-py27/9c15196/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/check/openstack-tox-py36/8812658/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/check/openstack-tox-py37/9c81db6/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/check/openstack-tox-docs/4c41978/html/ : SUCCESS in 4m 48s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-rabbit/e46d7f3/ : FAILURE in 4m 31s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-amqp1/41b4141/ : SUCCESS in 3m 55s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-kafka/11fbbf0/ : FAILURE in 4m 46s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-rabbit/6bb9960/ : SUCCESS in 4m 46s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-amqp1/ae9ed42/ : SUCCESS in 4m 57s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-kafka/534809d/ : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-rabbit/9b7865f/ : SUCCESS in 1h 49m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/80ade07/ : SUCCESS in 1h 33m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/e226f02/ : SUCCESS in 2h 02m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/d3c8857/ : FAILURE in 18m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/addebd5/ : FAILURE in 23m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm/e326532/ : FAILURE in 49m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm-multinode/8008721/ : SUCCESS in 1h 12m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/d372b84/ : FAILURE in 46m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/525e2b7/ : FAILURE in 43m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/2956b31/ : FAILURE in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44f0551002cd085d708e2bda2908eaec2f23fa3a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-05-29 17:28:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56dae97aee3384de99971c9e8307c8226bd52818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:32:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/20/654820/1/check/tempest-full/349f5e7/ : SUCCESS in 1h 43m 51s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/check/tempest-full-py3/04b9092/ : SUCCESS in 1h 39m 46s\n- openstack-tox-cover http://logs.openstack.org/20/654820/1/check/openstack-tox-cover/7bd86d7/cover/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/check/openstack-tox-lower-constraints/52f07cf/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/check/openstack-tox-pep8/43b5218/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/check/openstack-tox-py27/1c1a56b/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/check/openstack-tox-py36/e88201b/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/check/openstack-tox-py37/797d9b4/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/check/openstack-tox-docs/209527d/html/ : SUCCESS in 4m 03s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-rabbit/a1f3e55/ : SUCCESS in 5m 56s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-amqp1/cafd77f/ : SUCCESS in 5m 44s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py27-func-kafka/ef3e828/ : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-rabbit/50bae77/ : SUCCESS in 6m 06s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-amqp1/66cbbb6/ : SUCCESS in 4m 25s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-tox-py35-func-kafka/719c850/ : FAILURE in 4m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-rabbit/e6de4eb/ : SUCCESS in 1h 42m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e5e5473/ : SUCCESS in 1h 44m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/6698db3/ : FAILURE in 1h 32m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/853501f/ : FAILURE in 20m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/0e8a58f/ : FAILURE in 25m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm/c05b6e8/ : SUCCESS in 56m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/20/654820/1/check/oslo.messaging-src-grenade-dsvm-multinode/354fe54/ : SUCCESS in 1h 11m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/2fd87ee/ : FAILURE in 42m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/fc34d03/ : FAILURE in 46m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/20/654820/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/b359d53/ : FAILURE in 24m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b603c55ae37277f11478c8acf5bcd1ef3487242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:32:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a31f5a35f6921a2093f505e0cac8b3ba1e5efce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 21:22:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"721e28c27821e6cbfd4b8fe025be57ac75f6b06e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 21:22:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/20/654820/1/gate/tempest-full/d07ce3b/ : SUCCESS in 1h 44m 11s\n- tempest-full-py3 http://logs.openstack.org/20/654820/1/gate/tempest-full-py3/7c53c10/ : SUCCESS in 1h 36m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/654820/1/gate/openstack-tox-lower-constraints/abe4eed/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/20/654820/1/gate/openstack-tox-pep8/2abcfad/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py27/3d851bf/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py36/5fd1474/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/20/654820/1/gate/openstack-tox-py37/0c41075/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/20/654820/1/gate/openstack-tox-docs/858c455/html/ : SUCCESS in 3m 15s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-tox-py27-func-rabbit/a65c8d9/ : SUCCESS in 5m 09s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/20/654820/1/gate/oslo.messaging-src-dsvm-full-rabbit/2512c3c/ : SUCCESS in 1h 35m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99b77c07bee7dc482af03d13b2c454593f153f0c","revisions":{"99b77c07bee7dc482af03d13b2c454593f153f0c":{"kind":"REWORK","_number":1,"created":"2019-04-23 07:51:00.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/20/654820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/20/654820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/654820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/654820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/654820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/20/654820/1"}}},"commit":{"parents":[{"commit":"f276f0853afcda949a665d2fa12d712e96d7115f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f276f0853afcda949a665d2fa12d712e96d7115f"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-04-04 12:56:25.000000000","tz":120},"committer":{"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","date":"2019-04-23 07:51:00.000000000","tz":0},"subject":"Retry to declare a queue after internal error","message":"Retry to declare a queue after internal error\n\nWithout this commit, the client can lose the messages, because the\nclient does not handler the \u0027AMQP internal error 541\u0027,\nread here [2] for details.\nThe fix retries to create the queue after a delay.\nWhen the virtual-host is ready the declare does not fail.\nThis is a rare condiction, please read the bug [1] for details.\n\nCloses-Bug: #1822778\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1822778\n[2] https://www.rabbitmq.com/amqp-0-9-1-reference.html\n\nChange-Id: I7ab1f9d21ebb807285bf1422bc14cc6e07dcd32a\n(cherry picked from commit 4d2787227b00b973973554f7387e621d2664c0d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99b77c07bee7dc482af03d13b2c454593f153f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99b77c07bee7dc482af03d13b2c454593f153f0c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
