)]}'
{"id":"openstack%2Foslo.messaging~655462","triplet_id":"openstack%2Foslo.messaging~master~Ie82098a7dd2a084dfe4f1e077d1c84f270c7c051","project":"openstack/oslo.messaging","branch":"master","topic":"test-do-not-merge","hashtags":[],"change_id":"Ie82098a7dd2a084dfe4f1e077d1c84f270c7c051","subject":"Do Not Merge","status":"ABANDONED","created":"2019-04-24 14:41:23.000000000","updated":"2019-04-25 16:57:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48c1d4b23f2548568f407fdbf215ce1f02e6a346","_number":655462,"virtual_id_number":655462,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-04-24 17:05:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 17:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7f5b33c7a0eb834282b6f064f34d00986d77bac","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-24 14:41:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c56282dc7d281df23869b3c8b2d4ee1c192fc29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 17:05:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/62/655462/1/check/tempest-full/3f69cdd/ : SUCCESS in 1h 50m 21s\n- tempest-full-py3 http://logs.openstack.org/62/655462/1/check/tempest-full-py3/92d4f41/ : SUCCESS in 1h 41m 55s\n- openstack-tox-cover http://logs.openstack.org/62/655462/1/check/openstack-tox-cover/9356e2e/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/655462/1/check/openstack-tox-lower-constraints/c4eede7/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/62/655462/1/check/openstack-tox-pep8/3946a76/ : FAILURE in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/62/655462/1/check/openstack-tox-py27/f9270da/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/62/655462/1/check/openstack-tox-py36/4a19355/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/62/655462/1/check/openstack-tox-py37/ecafadc/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/62/655462/1/check/openstack-tox-docs/b1d8eab/html/ : SUCCESS in 3m 26s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py27-func-rabbit/0a0b97d/ : SUCCESS in 5m 23s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py27-func-amqp1/b0eb643/ : FAILURE in 11m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py27-func-kafka/171b583/ : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py36-func-rabbit/9ad1577/ : SUCCESS in 5m 46s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py36-func-amqp1/a87a08d/ : FAILURE in 12m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/62/655462/1/check/oslo.messaging-tox-py36-func-kafka/4147373/ : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-dsvm-full-rabbit/c12d471/ : SUCCESS in 1h 53m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/d348cfd/ : SUCCESS in 1h 41m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/0818c3d/ : SUCCESS in 2h 05m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/40f3365/ : SUCCESS in 1h 31m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/84fc228/ : SUCCESS in 1h 53m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-grenade-dsvm/63d86d8/ : SUCCESS in 52m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/62/655462/1/check/oslo.messaging-src-grenade-dsvm-multinode/f33265c/ : SUCCESS in 1h 25m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/62/655462/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/0cc991e/ : FAILURE in 1h 03m 27s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/62/655462/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/0c0d4b5/ : FAILURE in 51m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/62/655462/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/dda79b8/ : FAILURE in 1h 00m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"738a7638fea737f7ed9143a8e21fc12753afc129","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-25 16:57:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0dc0b06c5aedcbec9de0be9b934c6cc9d2987c4b","revisions":{"0dc0b06c5aedcbec9de0be9b934c6cc9d2987c4b":{"kind":"REWORK","_number":1,"created":"2019-04-24 14:41:23.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/62/655462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/62/655462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/655462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/655462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/655462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/62/655462/1"}}},"commit":{"parents":[{"commit":"8897e1584fcd977de29d989bf9d4932d374a5d4f","subject":"Update messaging intermediaries for amqp1 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8897e1584fcd977de29d989bf9d4932d374a5d4f"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-04-24 14:41:15.000000000","tz":-240},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-04-24 14:41:15.000000000","tz":-240},"subject":"Do Not Merge","message":"Do Not Merge\n\nChange-Id: Ie82098a7dd2a084dfe4f1e077d1c84f270c7c051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0dc0b06c5aedcbec9de0be9b934c6cc9d2987c4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0dc0b06c5aedcbec9de0be9b934c6cc9d2987c4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
