)]}'
{"id":"openstack%2Foslo.messaging~655737","triplet_id":"openstack%2Foslo.messaging~master~I8ae41eb3977fa273f1e2bb8b53b974d01310611a","project":"openstack/oslo.messaging","branch":"master","topic":"test-do-not-merge","hashtags":[],"change_id":"I8ae41eb3977fa273f1e2bb8b53b974d01310611a","subject":"Do Not Merge","status":"ABANDONED","created":"2019-04-25 16:58:38.000000000","updated":"2019-05-17 13:07:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"892ff95a5b3e8980901c3a2580eb9014807dc75d","_number":655737,"virtual_id_number":655737,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-04-25 19:22:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 19:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"083ad18207f06f3cea7668a378c2c2c3a8b78bcb","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-04-25 16:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ebef953a44d277006c3207839783b1b42f86739","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 19:22:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/37/655737/1/check/tempest-full/d1a1f80/ : TIMED_OUT in 2h 07m 57s\n- tempest-full-py3 http://logs.openstack.org/37/655737/1/check/tempest-full-py3/266face/ : SUCCESS in 1h 50m 47s\n- openstack-tox-cover http://logs.openstack.org/37/655737/1/check/openstack-tox-cover/9a8e252/cover/ : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/37/655737/1/check/openstack-tox-lower-constraints/ac43798/ : SUCCESS in 7m 15s\n- openstack-tox-pep8 http://logs.openstack.org/37/655737/1/check/openstack-tox-pep8/d7976db/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/37/655737/1/check/openstack-tox-py27/b4f3eaf/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/37/655737/1/check/openstack-tox-py36/6a452a6/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/37/655737/1/check/openstack-tox-py37/c090a31/ : SUCCESS in 5m 40s\n- openstack-tox-docs http://logs.openstack.org/37/655737/1/check/openstack-tox-docs/6bed90e/html/ : SUCCESS in 3m 40s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py27-func-rabbit/5b5661b/ : SUCCESS in 5m 39s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py27-func-amqp1/6cdbd4c/ : FAILURE in 12m 20s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py27-func-kafka/c69e212/ : FAILURE in 4m 46s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py36-func-rabbit/2f6a49e/ : SUCCESS in 5m 47s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py36-func-amqp1/09123b7/ : FAILURE in 11m 59s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/37/655737/1/check/oslo.messaging-tox-py36-func-kafka/52eb672/ : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-dsvm-full-rabbit/16a82ff/ : SUCCESS in 1h 53m 46s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e16d2e4/ : SUCCESS in 2h 08m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/b9b88ab/ : SUCCESS in 2h 09m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/e35e524/ : SUCCESS in 1h 58m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/6782685/ : SUCCESS in 1h 53m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-grenade-dsvm/30545b9/ : SUCCESS in 51m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/37/655737/1/check/oslo.messaging-src-grenade-dsvm-multinode/3762744/ : SUCCESS in 1h 14m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/37/655737/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/ff6bd8a/ : FAILURE in 1h 01m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/37/655737/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/7ecb3a3/ : FAILURE in 1h 07m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/37/655737/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/f508968/ : FAILURE in 1h 07m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58f52193809064e0f6c10d3f8391151f460b4f14","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-17 13:07:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4577d1030e43cc5b8024ce6c0a4c92264014abdd","revisions":{"4577d1030e43cc5b8024ce6c0a4c92264014abdd":{"kind":"REWORK","_number":1,"created":"2019-04-25 16:58:38.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/37/655737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/655737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/655737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/655737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/655737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/655737/1"}}},"commit":{"parents":[{"commit":"472d483942d77ff7157db159606f7e0599f2b6c6","subject":"Merge \"Change python3.5 job to python3.7 job on Stein+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/472d483942d77ff7157db159606f7e0599f2b6c6"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-04-25 16:57:44.000000000","tz":-240},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-04-25 16:57:44.000000000","tz":-240},"subject":"Do Not Merge","message":"Do Not Merge\n\nChange-Id: I8ae41eb3977fa273f1e2bb8b53b974d01310611a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4577d1030e43cc5b8024ce6c0a4c92264014abdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4577d1030e43cc5b8024ce6c0a4c92264014abdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
