)]}'
{"id":"openstack%2Foslo.messaging~658988","triplet_id":"openstack%2Foslo.messaging~master~I0eacd47e99515846884b5185b84f6ca8f525550b","project":"openstack/oslo.messaging","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"I0eacd47e99515846884b5185b84f6ca8f525550b","subject":"Uncap Bandit","status":"ABANDONED","created":"2019-05-14 08:56:27.000000000","updated":"2019-05-22 14:18:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec8b12aac122629951cd787292231fdf231f5193","_number":658988,"virtual_id_number":658988,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-14 11:09:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2019-05-15 15:49:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 11:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-15 15:49:46.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"}],"messages":[{"id":"d6fbfc24dbd0ab47c9ac1bbe243f2ef861369f9f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-14 08:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e058ab0c4b9b59ba486e05d3b2ad553f653e251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/88/658988/1/check/requirements-check/3ec5094/ : SUCCESS in 3m 44s\n- tempest-full http://logs.openstack.org/88/658988/1/check/tempest-full/7f54c43/ : SUCCESS in 1h 39m 50s\n- tempest-full-py3 http://logs.openstack.org/88/658988/1/check/tempest-full-py3/c25ce6b/ : SUCCESS in 1h 39m 00s\n- openstack-tox-cover http://logs.openstack.org/88/658988/1/check/openstack-tox-cover/a1a0909/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/658988/1/check/openstack-tox-lower-constraints/2b445aa/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/88/658988/1/check/openstack-tox-pep8/4aff31d/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/88/658988/1/check/openstack-tox-py27/0da0717/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/88/658988/1/check/openstack-tox-py36/04b6f0e/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/88/658988/1/check/openstack-tox-py37/bbc07ee/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/88/658988/1/check/openstack-tox-docs/633ee4c/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/88/658988/1/check/build-openstack-releasenotes/5608bf2/html/ : SUCCESS in 3m 27s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py27-func-rabbit/e8dc43a/ : SUCCESS in 5m 00s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py27-func-amqp1/c062fcd/ : FAILURE in 12m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py27-func-kafka/04bb280/ : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py36-func-rabbit/713cd5f/ : SUCCESS in 5m 38s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py36-func-amqp1/e6a708b/ : FAILURE in 11m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/88/658988/1/check/oslo.messaging-tox-py36-func-kafka/7ef7b25/ : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/88/658988/1/check/oslo.messaging-src-dsvm-full-rabbit/49f7b5c/ : SUCCESS in 1h 41m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/88/658988/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e14b789/ : SUCCESS in 1h 53m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/88/658988/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/41cc8de/ : SUCCESS in 2h 07m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/88/658988/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/d97cd57/ : SUCCESS in 1h 40m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/88/658988/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/853c47c/ : SUCCESS in 2h 03m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/88/658988/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/626e6ba/ : FAILURE in 46m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/88/658988/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/0a39cd7/ : FAILURE in 46m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/88/658988/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/db1234b/ : FAILURE in 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c22b417b2ba45b4a29f9da258342f1cf9add10f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-22 14:18:11.000000000","message":"Abandoned\n\nAbandoning per http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006432.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c5aa928ffcf3d5311aca8ba3dcc93142c4805d5","revisions":{"6c5aa928ffcf3d5311aca8ba3dcc93142c4805d5":{"kind":"REWORK","_number":1,"created":"2019-05-14 08:56:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/88/658988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/658988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/658988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/658988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/658988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/658988/1"}}},"commit":{"parents":[{"commit":"4f467fbd73092aadda8b42e517bb60a4782064ae","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f467fbd73092aadda8b42e517bb60a4782064ae"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-14 08:55:59.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-14 08:55:59.000000000","tz":120},"subject":"Uncap Bandit","message":"Uncap Bandit\n\nThis was capped due to a problem with how it processed exclusions.\nOnce that problem is fixed we want to uncap it and exlude only the\nspecific versions that are broken.\n\nChange-Id: I0eacd47e99515846884b5185b84f6ca8f525550b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c5aa928ffcf3d5311aca8ba3dcc93142c4805d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c5aa928ffcf3d5311aca8ba3dcc93142c4805d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
