)]}'
{"id":"openstack%2Foslo.messaging~659078","triplet_id":"openstack%2Foslo.messaging~master~Ic8f87208a0891dd21e99530ad0cf5caf22d2e785","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1661510","hashtags":[],"change_id":"Ic8f87208a0891dd21e99530ad0cf5caf22d2e785","subject":"WIP - Add the mandatory flag","status":"ABANDONED","created":"2019-05-14 14:31:17.000000000","updated":"2019-05-21 12:43:25.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d943979e8e70e93f26c2702ebc430a4da0aea1ab","_number":659078,"virtual_id_number":659078,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":-1,"date":"2019-05-14 15:46:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":-1,"date":"2019-05-17 14:20:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"date":"2019-05-14 14:36:50.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":-1,"date":"2019-05-17 14:12:03.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"date":"2019-05-14 14:36:14.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 14:36:14.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-05-14 14:36:50.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 15:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-17 14:12:03.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"}],"messages":[{"id":"a76548c2a26fa5e5bfed45bda48f1acc57d210f6","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-05-14 14:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5d9b80684e78c3c8a7e9ed3c1d8dbe3cdd83f9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 15:46:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/78/659078/1/check/tempest-full/d41a4f6/ : FAILURE in 28m 22s\n- tempest-full-py3 http://logs.openstack.org/78/659078/1/check/tempest-full-py3/8ee0d44/ : FAILURE in 29m 35s\n- openstack-tox-cover http://logs.openstack.org/78/659078/1/check/openstack-tox-cover/f3975b5/ : FAILURE in 8m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/78/659078/1/check/openstack-tox-lower-constraints/892d2e4/ : FAILURE in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/78/659078/1/check/openstack-tox-pep8/a049334/ : FAILURE in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/78/659078/1/check/openstack-tox-py27/d90869a/ : FAILURE in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/78/659078/1/check/openstack-tox-py36/a28b1d7/ : FAILURE in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/78/659078/1/check/openstack-tox-py37/6ff9904/ : FAILURE in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/78/659078/1/check/openstack-tox-docs/53b2dbe/html/ : SUCCESS in 3m 49s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py27-func-rabbit/d93fd3e/ : FAILURE in 6m 21s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py27-func-amqp1/5e17cbd/ : FAILURE in 12m 37s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py27-func-kafka/78d7a20/ : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py36-func-rabbit/2f299cb/ : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py36-func-amqp1/5f9db2d/ : FAILURE in 11m 10s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/78/659078/1/check/oslo.messaging-tox-py36-func-kafka/58ebb21/ : FAILURE in 4m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-dsvm-full-rabbit/f981d21/ : FAILURE in 35m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/89bd691/ : FAILURE in 30m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/67644f4/ : FAILURE in 43m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/5d2896c/ : FAILURE in 32m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/9481cc6/ : FAILURE in 38m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-grenade-dsvm/ec6612f/ : FAILURE in 50m 37s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/78/659078/1/check/oslo.messaging-src-grenade-dsvm-multinode/da7984f/ : FAILURE in 1h 09m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/78/659078/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/561686a/ : FAILURE in 52m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/78/659078/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/fad559c/ : FAILURE in 43m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/78/659078/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/02b97a7/ : FAILURE in 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbf4b9f64835781248b508a38ee01d50116ab239","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-17 14:12:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a40480396c852063628b221bcef5484c2b0fea63","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-17 14:20:38.000000000","message":"Patch Set 1: Code-Review-1\n\nHi Gabriele,\n\nThanks for the patch - this will be a useful feature.\n\nHowever I gave you some bad advice when I suggested we make this feature configurable per call.  We can\u0027t do that since all call arguments are application defined (and transparent to oslo.messaging).  Adding the option parameter will break stuff badly.\n\nAgain, my bad - sorry.\n\nSo if we take a step back this feature will have to be implemented as a new parameter to the __init__ of the RPCClient class.\n\nThat should make the implementation easier - you won\u0027t have to monkey around with the message envelope and all that.\n\nTake a look at the way we pass \u0027call_monitor_timeout\u0027 as an example - you\u0027ll do the same to get the flag down into the amqpdriver.py/impl_rabbit.py code.\n\nMost of the other stuff looks good so far.  One thing you\u0027ll need to do is avoid adding all that whitespace - it\u0027s unnecessary and makes the patch hard to read (perhaps your editor is adding that?).","accounts_in_message":[],"_revision_number":1},{"id":"52d4c5737512a7940937b590a92df93596f15b5b","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-17 14:25:38.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"59fc1b779f739519240e48e887207d0bb40be253","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-05-21 12:43:25.000000000","message":"Abandoned\n\nclosed in favor of https://review.opendev.org/#/c/660373/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08aa3a4e8561b05066f679a697fc571d7d0744dd","revisions":{"08aa3a4e8561b05066f679a697fc571d7d0744dd":{"kind":"REWORK","_number":1,"created":"2019-05-14 14:31:17.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/78/659078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/659078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/659078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/659078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/659078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/659078/1"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-14 13:33:57.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-14 14:31:15.000000000","tz":120},"subject":"WIP - Add the mandatory flag","message":"WIP - Add the mandatory flag\n\nAdd mandatory flag to handle the no-routed messages.\nIn this way it is possible to know if a message reaches a queue or not.\nIt raises an exception in case a message is not routed.\n\nNote:\n- This is a WIP\n- This is just a an idea, I wanted to add a generic parameter called\n\"option\", so we can add more options during the publish.\nSee the gist[1] for details.\n\nCloses-bug: #1661510\n\n1- https://gist.github.com/Gsantomaggio/bff62fe28210622cfa9d32d164412800\n\nChange-Id: Ic8f87208a0891dd21e99530ad0cf5caf22d2e785\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/08aa3a4e8561b05066f679a697fc571d7d0744dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/08aa3a4e8561b05066f679a697fc571d7d0744dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
