)]}'
{"id":"openstack%2Foslo.messaging~660277","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I9dc1644cac0e39eb11bf05f57bde77dcf6d42ed3","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"bug/1828841","hashtags":[],"change_id":"I9dc1644cac0e39eb11bf05f57bde77dcf6d42ed3","subject":"Fix switch connection destination when a rabbitmq cluster node disappear","status":"MERGED","created":"2019-05-21 08:17:57.000000000","updated":"2019-06-06 11:03:11.000000000","submitted":"2019-06-06 11:03:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660277-1559818991554-3cf4fd82","meta_rev_id":"0343ec9c399adfcc8b5a93c1b744995972923d7c","_number":660277,"virtual_id_number":660277,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-05-23 12:57:41.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-06-06 11:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-24 14:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-06 09:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-06 09:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 14:06:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-06 09:07:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-06 11:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"569a191ff9c2a9379d063f45b570346e13065fc8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 08:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0bcea3a13e0e9f3e83d41f1f6511fc8fb0ec092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:10:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/77/660277/1/check/tempest-full/a5e0ca3/ : SUCCESS in 1h 47m 12s\n- tempest-full-py3 http://logs.openstack.org/77/660277/1/check/tempest-full-py3/129d1d3/ : SUCCESS in 1h 49m 50s\n- openstack-tox-cover http://logs.openstack.org/77/660277/1/check/openstack-tox-cover/0edb89c/cover/ : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/660277/1/check/openstack-tox-lower-constraints/d46160a/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/77/660277/1/check/openstack-tox-pep8/cb85f6d/ : FAILURE in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/77/660277/1/check/openstack-tox-py27/9a5515c/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/77/660277/1/check/openstack-tox-py36/9451d2c/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/77/660277/1/check/openstack-tox-py37/5c6ddd7/ : SUCCESS in 5m 28s\n- openstack-tox-docs http://logs.openstack.org/77/660277/1/check/openstack-tox-docs/11cff39/html/ : SUCCESS in 3m 13s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py27-func-rabbit/13f954e/ : SUCCESS in 4m 42s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py27-func-amqp1/a565145/ : SUCCESS in 4m 32s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py27-func-kafka/3caf4f7/ : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py35-func-rabbit/9536cc2/ : FAILURE in 5m 31s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py35-func-amqp1/e10ba9c/ : SUCCESS in 4m 47s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/77/660277/1/check/oslo.messaging-tox-py35-func-kafka/092e33c/ : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-dsvm-full-rabbit/4400c2c/ : SUCCESS in 1h 45m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/7c4b692/ : SUCCESS in 1h 49m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/63fe4a6/ : FAILURE in 1h 46m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/d644ff7/ : FAILURE in 18m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/54fce68/ : FAILURE in 20m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-grenade-dsvm/603342a/ : SUCCESS in 57m 46s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/77/660277/1/check/oslo.messaging-src-grenade-dsvm-multinode/554957a/ : SUCCESS in 1h 08m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/77/660277/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/6a51f58/ : FAILURE in 44m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/77/660277/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/485cf16/ : FAILURE in 46m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/77/660277/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/8e12791/ : FAILURE in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc880ab333f7b561c273d211ff0b75628039206e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 11:35:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dfd5dcf7522dee57b63ded6d27b4d9fa753b0cc0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 13:31:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0b97115e0245cc65405191eab69506753684584e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:31:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/77/660277/2/check/tempest-full/aa3bc6c/ : SUCCESS in 1h 43m 21s\n- tempest-full-py3 http://logs.openstack.org/77/660277/2/check/tempest-full-py3/326a46f/ : SUCCESS in 1h 35m 16s\n- openstack-tox-cover http://logs.openstack.org/77/660277/2/check/openstack-tox-cover/a99ca92/cover/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/660277/2/check/openstack-tox-lower-constraints/0fe91e4/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/77/660277/2/check/openstack-tox-pep8/4badbf0/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/77/660277/2/check/openstack-tox-py27/85a9b6f/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/77/660277/2/check/openstack-tox-py36/6c38ec0/ : FAILURE in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/77/660277/2/check/openstack-tox-py37/8ce54c6/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/77/660277/2/check/openstack-tox-docs/cd070c2/html/ : SUCCESS in 3m 07s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-rabbit/240db8f/ : SUCCESS in 5m 08s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-amqp1/b759ae8/ : SUCCESS in 4m 08s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-kafka/29d346a/ : FAILURE in 3m 58s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-rabbit/227afa1/ : FAILURE in 4m 27s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-amqp1/d4a9bcb/ : SUCCESS in 5m 30s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-kafka/35a1812/ : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-rabbit/7b18bed/ : SUCCESS in 1h 56m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/462840c/ : SUCCESS in 1h 35m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/4a20bbc/ : FAILURE in 1h 37m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/6d25e59/ : FAILURE in 18m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/62837a4/ : FAILURE in 28m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-grenade-dsvm/d4c940d/ : SUCCESS in 55m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-grenade-dsvm-multinode/05b5131/ : SUCCESS in 1h 08m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/9760767/ : FAILURE in 42m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/4bdf48e/ : FAILURE in 53m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/55914de/ : FAILURE in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93575b57b334cb10f6451aff2add6e68e340e240","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-23 12:57:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31ed87c6d7d049b9eb51025026d4cb930ac6a4fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 14:57:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/77/660277/2/check/tempest-full/775a06c/ : SUCCESS in 1h 57m 05s\n- tempest-full-py3 http://logs.openstack.org/77/660277/2/check/tempest-full-py3/f982d0d/ : SUCCESS in 1h 42m 02s\n- openstack-tox-cover http://logs.openstack.org/77/660277/2/check/openstack-tox-cover/d4fc857/cover/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/660277/2/check/openstack-tox-lower-constraints/12fd08c/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/77/660277/2/check/openstack-tox-pep8/0d942f4/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/77/660277/2/check/openstack-tox-py27/b691e45/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/77/660277/2/check/openstack-tox-py36/a131a9e/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/77/660277/2/check/openstack-tox-py37/2dc56a5/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/77/660277/2/check/openstack-tox-docs/3a64698/html/ : SUCCESS in 2m 46s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-rabbit/5dfeb39/ : SUCCESS in 4m 48s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-amqp1/21b5c29/ : FAILURE in 3m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py27-func-kafka/fb3de02/ : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-rabbit/d857422/ : FAILURE in 5m 31s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-amqp1/2b7aecf/ : SUCCESS in 4m 32s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-tox-py35-func-kafka/3f1f388/ : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-rabbit/343b55a/ : SUCCESS in 1h 28m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/b3422a7/ : SUCCESS in 1h 43m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/45c0907/ : SUCCESS in 1h 51m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/b1d8e09/ : FAILURE in 17m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5a1ba50/ : FAILURE in 24m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-grenade-dsvm/b9e317a/ : SUCCESS in 59m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/77/660277/2/check/oslo.messaging-src-grenade-dsvm-multinode/b90b490/ : SUCCESS in 1h 11m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/88b197f/ : FAILURE in 45m 40s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/9ec92c0/ : FAILURE in 46m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/77/660277/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/d169ddc/ : FAILURE in 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7172c8ec50fd540d1928424bfda71cfdb785bef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-24 14:06:52.000000000","message":"Patch Set 2: Code-Review+2\n\nMerged to master, good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"64315cc0efb69aea58de4539c5115a8b016e4c61","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-06 09:07:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean backport. A-ok for me","accounts_in_message":[],"_revision_number":2},{"id":"587ba1db4fd367672552d717380e557c0e3296fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:07:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf8199744ce9d1744a953688a8d01498614a8794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:03:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a77e54815dec2fcb83e73899ff388dc22a162dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:03:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/77/660277/2/gate/tempest-full/9e64127/ : SUCCESS in 1h 54m 18s\n- tempest-full-py3 http://logs.openstack.org/77/660277/2/gate/tempest-full-py3/0aca23b/ : SUCCESS in 1h 23m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/660277/2/gate/openstack-tox-lower-constraints/c562164/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/77/660277/2/gate/openstack-tox-pep8/9eea6ce/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/77/660277/2/gate/openstack-tox-py27/38db72d/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/77/660277/2/gate/openstack-tox-py36/b21cf16/ : SUCCESS in 5m 02s\n- openstack-tox-py37 http://logs.openstack.org/77/660277/2/gate/openstack-tox-py37/d38c97b/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/77/660277/2/gate/openstack-tox-docs/d334902/html/ : SUCCESS in 3m 22s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/77/660277/2/gate/oslo.messaging-tox-py27-func-rabbit/ae3f566/ : SUCCESS in 6m 13s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/77/660277/2/gate/oslo.messaging-src-dsvm-full-rabbit/be65b0f/ : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b8fd6370c8b04c3836ee4f9d06eaa90c7be5197","revisions":{"4f35fa2a15c47fd19d362c78047c758b72b68b1d":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:17:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/77/660277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/77/660277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/1"}}},"commit":{"parents":[{"commit":"f276f0853afcda949a665d2fa12d712e96d7115f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f276f0853afcda949a665d2fa12d712e96d7115f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-02 22:55:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-21 08:17:38.000000000","tz":120},"subject":"Fix switch connection destination when a rabbitmq cluster node disappear","message":"Fix switch connection destination when a rabbitmq cluster node disappear\n\nIn a clustered rabbitmq when a node disappears, we get a\nConnectionRefusedError because the socket get disconnected.\n\nThe socket access yields a OSError because the heartbeat\ntries to reach an unreachable host (No route to host).\n\nCatch these exceptions to ensure that we call ensure_connection for switching\nthe connection destination.\n\nPOC is available at github.com:4383/rabbitmq-oslo_messging-error-poc\n\nExample:\n    $ git clone git@github.com:4383/rabbitmq-oslo_messging-error-poc\n    $ cd rabbitmq-oslo_messging-error-poc\n    $ python -m virtualenv .\n    $ source bin/activate\n    $ pip install -r requirements.txt\n    $ sudo podman run -d --hostname my-rabbit --name rabbit rabbitmq:3\n    $ python poc.py $(sudo podman inspect rabbit | niet \u0027.[0].NetworkSettings.IPAddress\u0027)\n\nAnd in parallele in an another shell|tmux\n    $ podman stop rabbit\n    $ # observe the output of the poc.py script we now call ensure_connection\n\nNow you can observe some output relative to the connection who is\nmodified and not catched before these changes.\n\nRelated to: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1665399\n\nCloses-Bug: #1828841\n\nChange-Id: I9dc1644cac0e39eb11bf05f57bde77dcf6d42ed3\n(cherry picked from commit 9d8b1430e5c081b081c0e3c0b5f12f744dc7809d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f35fa2a15c47fd19d362c78047c758b72b68b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f35fa2a15c47fd19d362c78047c758b72b68b1d"}]},"branch":"refs/heads/stable/stein"},"7b8fd6370c8b04c3836ee4f9d06eaa90c7be5197":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-21 13:31:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/77/660277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/77/660277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/77/660277/2"}}},"commit":{"parents":[{"commit":"0f91b0a2115f3c0a817abb624d65f8f0dcad40b2","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f91b0a2115f3c0a817abb624d65f8f0dcad40b2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-02 22:55:56.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-21 13:24:23.000000000","tz":120},"subject":"Fix switch connection destination when a rabbitmq cluster node disappear","message":"Fix switch connection destination when a rabbitmq cluster node disappear\n\nIn a clustered rabbitmq when a node disappears, we get a\nConnectionRefusedError because the socket get disconnected.\n\nThe socket access yields a OSError because the heartbeat\ntries to reach an unreachable host (No route to host).\n\nCatch these exceptions to ensure that we call ensure_connection for switching\nthe connection destination.\n\nPOC is available at github.com:4383/rabbitmq-oslo_messging-error-poc\n\nExample:\n    $ git clone git@github.com:4383/rabbitmq-oslo_messging-error-poc\n    $ cd rabbitmq-oslo_messging-error-poc\n    $ python -m virtualenv .\n    $ source bin/activate\n    $ pip install -r requirements.txt\n    $ sudo podman run -d --hostname my-rabbit --name rabbit rabbitmq:3\n    $ python poc.py $(sudo podman inspect rabbit | niet \u0027.[0].NetworkSettings.IPAddress\u0027)\n\nAnd in parallele in an another shell|tmux\n    $ podman stop rabbit\n    $ # observe the output of the poc.py script we now call ensure_connection\n\nNow you can observe some output relative to the connection who is\nmodified and not catched before these changes.\n\nRelated to: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1665399\n\nCloses-Bug: #1828841\n\nChange-Id: I9dc1644cac0e39eb11bf05f57bde77dcf6d42ed3\n(cherry picked from commit 9d8b1430e5c081b081c0e3c0b5f12f744dc7809d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7b8fd6370c8b04c3836ee4f9d06eaa90c7be5197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7b8fd6370c8b04c3836ee4f9d06eaa90c7be5197"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
