)]}'
{"id":"openstack%2Foslo.messaging~660373","triplet_id":"openstack%2Foslo.messaging~master~Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72","project":"openstack/oslo.messaging","branch":"master","topic":"bp/transport-options","hashtags":[],"change_id":"Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72","subject":"Add transport_options parameter","status":"MERGED","created":"2019-05-21 12:42:10.000000000","updated":"2020-05-12 06:08:35.000000000","submitted":"2019-06-17 16:35:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660373-1560789355733-d473cd44","meta_rev_id":"d36252df766a1e992f62b745f214b5d0a79d4b68","_number":660373,"virtual_id_number":660373,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2019-06-17 12:27:31.000000000","_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-06-17 16:35:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-06-15 12:04:04.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":2,"date":"2019-06-17 07:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":1,"date":"2019-06-17 07:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 13:03:11.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-05-28 19:35:03.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-05-29 11:15:31.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-06-07 04:11:49.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-06-11 16:50:55.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-06-11 21:34:33.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-15 12:04:04.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-06-17 07:47:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-06-17 16:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a31780b783edd0bed35ade9b01b6d8c5fc8c941","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-05-21 12:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aaf916a6d5ed00f845c4277767defe4698675ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:53:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/1/check/tempest-full/0c1c973/ : SUCCESS in 1h 42m 35s\n- tempest-full-py3 http://logs.openstack.org/73/660373/1/check/tempest-full-py3/e8e06e2/ : SUCCESS in 1h 35m 43s\n- openstack-tox-cover http://logs.openstack.org/73/660373/1/check/openstack-tox-cover/6b291f3/ : FAILURE in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/1/check/openstack-tox-lower-constraints/072a55e/ : FAILURE in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/1/check/openstack-tox-pep8/14980c8/ : FAILURE in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/1/check/openstack-tox-py27/c3c96bf/ : FAILURE in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/1/check/openstack-tox-py36/16966cc/ : FAILURE in 5m 02s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/1/check/openstack-tox-py37/e728251/ : FAILURE in 7m 10s\n- openstack-tox-docs http://logs.openstack.org/73/660373/1/check/openstack-tox-docs/d95bac7/html/ : SUCCESS in 3m 06s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py27-func-rabbit/2a354af/ : FAILURE in 4m 29s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py27-func-amqp1/c0c3639/ : FAILURE in 11m 43s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py27-func-kafka/888402b/ : FAILURE in 3m 34s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py36-func-rabbit/27aee0e/ : FAILURE in 4m 45s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py36-func-amqp1/4bee357/ : FAILURE in 11m 01s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/1/check/oslo.messaging-tox-py36-func-kafka/64f98a6/ : FAILURE in 4m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-dsvm-full-rabbit/0ff5ae0/ : SUCCESS in 1h 40m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/fd9e028/ : FAILURE in 29m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/d4cc46f/ : FAILURE in 40m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/4c80356/ : SUCCESS in 1h 44m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/2178690/ : SUCCESS in 1h 58m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-grenade-dsvm/fd0ce48/ : SUCCESS in 58m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/1/check/oslo.messaging-src-grenade-dsvm-multinode/bd112e4/ : SUCCESS in 1h 10m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/f5f9496/ : FAILURE in 49m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/517bbab/ : FAILURE in 40m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/eccb8a0/ : FAILURE in 51m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32534e5acf351f6d9ece1b4ca6f3d38a5aa05a51","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-27 18:13:30.000000000","message":"Patch Set 1: Code-Review+1\n\nAfter a quick look, I personaly love your proposed changes, I can easily imagine some use cases which can use retrying patterns with that for some reasons. \n\nBy example using tenacity [1] and retry on error, see the example bellow.\n\nExample on the customer side where retrying is needed:\n```\nfrom tenacity import *\nfrom oslo_messaging.exceptions import MessageUndeliverable\n\n@retry(retry\u003dretry_if_exception_type(MessageUndeliverable))\ndef call(transport, target, number):\n    client \u003d oslo_messaging.RPCClient(transport, target, options\u003d{\u0027mandatory\u0027: True})\n\n    for i in range(1, 50):\n        time.sleep(3)\n        r \u003d client.call({}, \u0027foo\u0027, id_value\u003dstr(i), test_value\u003d\"hello oslo\")\n        print(\"hello\" + r + \" - number: \" + str(number))\n```\n\nI need to do more tests and review it more deeply but globally I really love it :)\n\nI put a simple +1 for now to notice that I appreciate your changes even if it still WIP.\n\n[1] https://github.com/jd/tenacity","accounts_in_message":[],"_revision_number":1},{"id":"210ef8c7443da49b0aa40224f411defa4c242e7c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-29 11:15:31.000000000","message":"Patch Set 1: Code-Review+1\n\nI haven\u0027t tested this, but conceptually I really like the idea.","accounts_in_message":[],"_revision_number":1},{"id":"59f840d1f8c4f02becf094f5d9ccbb3588e171c9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-29 14:15:06.000000000","message":"Patch Set 1:\n\nI have to admit I\u0027m a little surprised kombu doesn\u0027t already raise an exception if a message can\u0027t be enqueued. I understand that it can\u0027t guarantee delivery, but if I ask to send something and it can\u0027t be sent for some reason that seems like a pretty obvious error.\n\nIn any case, +1 to this behavior. Still waiting on one of our messaging folks to comment, but I think they were onboard with this as well. Note that the WIP will need to be removed before this merges, and obviously the unit test failures will need to be fixed.","accounts_in_message":[],"_revision_number":1},{"id":"8c3cf6d5c86c04dd2a3967e8f195d1a8784f0f90","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-29 14:18:45.000000000","message":"Patch Set 1:\n\nOh, one other thing: What impact does this have on the driver interface? This option is specific to rabbit, will the other drivers need to start taking an \"options\" parameter too that will be ignored? Either way I guess we\u0027ll need to document what options a given driver accepts.","accounts_in_message":[],"_revision_number":1},{"id":"2fb3e30b81d5a1115f2210355aa642774794061f","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-05-29 14:32:15.000000000","message":"Patch Set 1:\n\nThanks Ben,\nRabbitMQ by default _does not raise any exception_ if the message is not routed. \nYou can test the mandatory flag in kombu following this example:\nhttps://github.com/celery/kombu/issues/1039 \n\nby setting mandatory\u003dTrue, then mandatory\u003dFalse.\n\nBy protocol the mandatory flag is False.\n\nThis is a WIP or (better) an idea, the option flag is something that we can use in all the drivers. \n\nI will add the documentation for sure, but the interface won\u0027t change since the options will be None","accounts_in_message":[],"_revision_number":1},{"id":"a78d2c1bcc0c331db6c9fe9a31d934c9c000648d","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-03 09:33:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e6c5aca84c6c9c85ae00bb024bfc5b84b223c0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 11:54:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/2/check/tempest-full/f38227c/ : POST_FAILURE in 1h 43m 30s\n- tempest-full-py3 http://logs.openstack.org/73/660373/2/check/tempest-full-py3/a7fc4ff/ : SUCCESS in 1h 40m 56s\n- openstack-tox-cover http://logs.openstack.org/73/660373/2/check/openstack-tox-cover/95165c6/ : FAILURE in 5m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/2/check/openstack-tox-lower-constraints/c0717ce/ : FAILURE in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/2/check/openstack-tox-pep8/701a638/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/2/check/openstack-tox-py27/ee42528/ : FAILURE in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/2/check/openstack-tox-py36/48a8621/ : FAILURE in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/2/check/openstack-tox-py37/6e5540d/ : FAILURE in 8m 07s\n- openstack-tox-docs http://logs.openstack.org/73/660373/2/check/openstack-tox-docs/e31dc58/html/ : SUCCESS in 2m 58s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-rabbit/6fc400e/ : FAILURE in 5m 54s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-amqp1/04f8492/ : FAILURE in 11m 46s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-kafka/8763ad1/ : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-rabbit/cdbd9c0/ : FAILURE in 5m 39s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-amqp1/bc854ce/ : FAILURE in 11m 11s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-kafka/6dd0baa/ : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-rabbit/04cd699/ : SUCCESS in 1h 39m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/30d4c4e/ : FAILURE in 28m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/f3c2784/ : FAILURE in 37m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/540eb26/ : SUCCESS in 1h 52m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/4c765e8/ : SUCCESS in 2h 08m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-grenade-dsvm/f795ea6/ : SUCCESS in 1h 03m 03s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-grenade-dsvm-multinode/de55656/ : SUCCESS in 1h 14m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/6c4ab26/ : FAILURE in 42m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/5c618fb/ : FAILURE in 41m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/76e9888/ : FAILURE in 47m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e99c02fe50e6fa2eabc5f80171279c7ea309040c","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-03 18:57:47.000000000","message":"Patch Set 2:\n\nThis review depends on Kombu issue 1050[1] \nI am waiting for the merge\n\n1 - https://github.com/celery/kombu/pull/1053","accounts_in_message":[],"_revision_number":2},{"id":"373c84e17d94bf7033ec7017dc46518d141fa890","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-04 16:16:38.000000000","message":"Patch Set 2:\n\nrecheck\n\n \u003e This review depends on Kombu issue 1050[1]\n \u003e I am waiting for the merge\n \u003e \n \u003e 1 - https://github.com/celery/kombu/pull/1053\n\nThe celery PR is now merged \u003d\u003e https://github.com/celery/kombu/pull/1053","accounts_in_message":[],"_revision_number":2},{"id":"2e31befb0d09486c7a980b832227fe651de8052e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 18:48:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/2/check/tempest-full/386e6c7/ : SUCCESS in 1h 50m 01s\n- tempest-full-py3 http://logs.openstack.org/73/660373/2/check/tempest-full-py3/0392862/ : SUCCESS in 1h 54m 28s\n- openstack-tox-cover http://logs.openstack.org/73/660373/2/check/openstack-tox-cover/3f6fccc/ : FAILURE in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/2/check/openstack-tox-lower-constraints/d45a4c7/ : FAILURE in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/2/check/openstack-tox-pep8/1f08cf1/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/2/check/openstack-tox-py27/63229a6/ : FAILURE in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/2/check/openstack-tox-py36/57423ad/ : FAILURE in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/2/check/openstack-tox-py37/689b51b/ : FAILURE in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/73/660373/2/check/openstack-tox-docs/29a362f/html/ : SUCCESS in 2m 57s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-rabbit/91585e1/ : FAILURE in 5m 13s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-amqp1/f89ef6d/ : FAILURE in 11m 19s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py27-func-kafka/ae6f91d/ : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-rabbit/0f11fd0/ : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-amqp1/9955c1a/ : FAILURE in 11m 01s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-tox-py36-func-kafka/0663de0/ : FAILURE in 3m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-rabbit/8bce0a2/ : SUCCESS in 1h 45m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/4e821ba/ : FAILURE in 31m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/d60cc00/ : FAILURE in 40m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/9ae032f/ : SUCCESS in 1h 48m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/54d58d7/ : SUCCESS in 2h 17m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-grenade-dsvm/6199bac/ : SUCCESS in 53m 45s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/2/check/oslo.messaging-src-grenade-dsvm-multinode/b1d33b3/ : SUCCESS in 1h 07m 25s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/33e7eaa/ : FAILURE in 47m 28s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/86ee08e/ : FAILURE in 42m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/1cfbd9c/ : FAILURE in 49m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d521709ded02c68097d8b8660d3e83ce06a26aaf","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-07 01:42:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36e8a9291c15c36beb2aef72a0a3b4a38585944f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-07 04:11:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b9ddf4c2f96c31ebb58cc29497c4b223f2642c8","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-07 07:48:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7fb043a74cbf6b2ec49ca4ccad0ebfd7aa1c1dc","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-10 11:10:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9805a5779874e5b68d690790cc3569a344d41eaa","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-10 11:12:53.000000000","message":"Patch Set 3:\n\nKombu 4.6.1 has been released [1] with my patch[2] for  `on_return` memory transport. \n\n\n1-https://github.com/celery/kombu/releases/tag/4.6.1\n2-https://github.com/celery/kombu/pull/1053","accounts_in_message":[],"_revision_number":3},{"id":"46bfb02ad5bc83aca894c52a5868edf0b57d00b4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-10 11:35:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7aa75a7de7f9233671b523727c5cce791ae03b22","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-10 12:09:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20b686faabd394883844ba6e74507d4f08b58206","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-10 12:51:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99e8b9f823e41aa48bcac8584fe4ff326b1599f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:50:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/73/660373/3/check/requirements-check/bb2ce74/ : FAILURE in 5m 17s\n- tempest-full http://logs.openstack.org/73/660373/3/check/tempest-full/53dd905/ : SUCCESS in 1h 38m 24s\n- tempest-full-py3 http://logs.openstack.org/73/660373/3/check/tempest-full-py3/4a8c19e/ : SUCCESS in 1h 32m 17s\n- openstack-tox-cover http://logs.openstack.org/73/660373/3/check/openstack-tox-cover/f88f85d/ : FAILURE in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/3/check/openstack-tox-lower-constraints/2db84bb/ : FAILURE in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/3/check/openstack-tox-pep8/bcf9e7c/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/3/check/openstack-tox-py27/06463a9/ : FAILURE in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/3/check/openstack-tox-py36/e869e86/ : FAILURE in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/3/check/openstack-tox-py37/d42d2c8/ : FAILURE in 5m 56s\n- openstack-tox-docs http://logs.openstack.org/73/660373/3/check/openstack-tox-docs/60beade/html/ : SUCCESS in 3m 22s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py27-func-rabbit/3c22f14/ : FAILURE in 5m 19s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py27-func-amqp1/8de3d94/ : FAILURE in 11m 40s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py27-func-kafka/24f74d4/ : FAILURE in 4m 39s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py36-func-rabbit/207f153/ : FAILURE in 5m 21s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py36-func-amqp1/bfddd68/ : FAILURE in 11m 45s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/3/check/oslo.messaging-tox-py36-func-kafka/9690954/ : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-dsvm-full-rabbit/e1d7726/ : SUCCESS in 1h 36m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/b73f288/ : FAILURE in 47m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/c68b0bd/ : FAILURE in 31m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3a49141/ : SUCCESS in 1h 45m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5989bb8/ : SUCCESS in 2h 03m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-grenade-dsvm/e5a3b4a/ : SUCCESS in 1h 00m 02s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/3/check/oslo.messaging-src-grenade-dsvm-multinode/dea7ebd/ : SUCCESS in 1h 09m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/5fb8d2c/ : FAILURE in 48m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/a5dd929/ : FAILURE in 44m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/5849b94/ : FAILURE in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d6c37c494d169537bb89ad32c5b9978b15d7c8a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-10 14:11:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15834c40d0eb83d118a6b4e0fccd8fbc577682bb","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-10 16:22:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c4e846437d154333249e6315df1461f331585d1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 19:16:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/73/660373/4/check/requirements-check/8352efa/ : SUCCESS in 3m 33s\n- tempest-full http://logs.openstack.org/73/660373/4/check/tempest-full/e26f478/ : SUCCESS in 1h 54m 37s\n- tempest-full-py3 http://logs.openstack.org/73/660373/4/check/tempest-full-py3/4f39031/ : SUCCESS in 1h 35m 08s\n- openstack-tox-cover http://logs.openstack.org/73/660373/4/check/openstack-tox-cover/a620fd3/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/4/check/openstack-tox-lower-constraints/7792954/ : SUCCESS in 5m 50s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/4/check/openstack-tox-pep8/935c13d/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/4/check/openstack-tox-py27/c4cfe56/ : SUCCESS in 7m 02s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/4/check/openstack-tox-py36/822f4f0/ : SUCCESS in 6m 35s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/4/check/openstack-tox-py37/e8cdfd1/ : SUCCESS in 6m 43s\n- openstack-tox-docs http://logs.openstack.org/73/660373/4/check/openstack-tox-docs/5c44a03/html/ : SUCCESS in 3m 40s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py27-func-rabbit/95d82e9/ : SUCCESS in 6m 25s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py27-func-amqp1/88cba63/ : FAILURE in 12m 20s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py27-func-kafka/4450339/ : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py36-func-rabbit/e4edf10/ : SUCCESS in 5m 53s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py36-func-amqp1/feb3e89/ : FAILURE in 12m 44s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/4/check/oslo.messaging-tox-py36-func-kafka/1cd0380/ : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-dsvm-full-rabbit/01bb768/ : SUCCESS in 1h 33m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/7a71fb8/ : FAILURE in 26m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/b42589e/ : FAILURE in 40m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-dsvm-full-kafka-hybrid/bf72265/ : SUCCESS in 1h 56m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-dsvm-full-kafka-centos-7/3cf08da/ : SUCCESS in 2h 10m 28s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-grenade-dsvm/f9652ab/ : SUCCESS in 1h 03m 50s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/4/check/oslo.messaging-src-grenade-dsvm-multinode/9f3aa6b/ : SUCCESS in 1h 13m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/4f3e5ad/ : FAILURE in 44m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c4c8eb3/ : FAILURE in 45m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/952e6a7/ : FAILURE in 55m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ef73c954b298824600717b858a92ee440bf0e6e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 19:44:04.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nOne thing to consider would be splitting out the \"options\" addition from the change to add mandatory to the options. I would expect we could merge the option framework change (without any options actually being set) pretty quickly and then limit the discussion on this patch to the behavior of the mandatory flag.\n\nIf it\u0027s a lot of work to disentangle the two then it\u0027s not worth it though. It\u0027s just a thought I had.","accounts_in_message":[],"_revision_number":3},{"id":"8d8a5503b6f56bcd01d81921a10e06b89be275c8","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-11 14:58:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"16072ce2bd5c57f04c7d96f59f5df0059c822143","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-11 15:04:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f70f84a5e5e7cc1c2829f0d1854cbffd148b0921","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-11 15:15:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a0ca0b0ed8825f5a5ae43c5551cb0851312c21df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-11 16:41:09.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM, the exception part have been removed so these changes doesn\u0027t introduce default behaviour that could be an issue for oslo.messaging users.","accounts_in_message":[],"_revision_number":7},{"id":"eaeb8cc9484ef32174cfe4c3d4f69a18668a3ded","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-11 16:50:55.000000000","message":"Patch Set 7: Code-Review+1\n\nok spliting this makes sense but im hoping we can start using the mandatory flag too however","accounts_in_message":[],"_revision_number":7},{"id":"a60caf0997b2860bfa7fcabcb51db46de4dfb3d1","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-11 17:20:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nA couple of nits on the parameter naming, and I don\u0027t think we want a release note for an internal change like this.","accounts_in_message":[],"_revision_number":7},{"id":"c007233a674a731fb2c7e73d9e9ac69b67b84a0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 17:34:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/7/check/tempest-full/afcf15b/ : TIMED_OUT in 2h 05m 40s\n- tempest-full-py3 http://logs.openstack.org/73/660373/7/check/tempest-full-py3/3fa5c7a/ : SUCCESS in 1h 54m 04s\n- openstack-tox-cover http://logs.openstack.org/73/660373/7/check/openstack-tox-cover/d25171a/cover/ : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/7/check/openstack-tox-lower-constraints/74cc06f/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/7/check/openstack-tox-pep8/a3d0715/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/7/check/openstack-tox-py27/8742f3c/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/7/check/openstack-tox-py36/cb1e09a/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/7/check/openstack-tox-py37/f10af78/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/73/660373/7/check/openstack-tox-docs/5f5065b/html/ : SUCCESS in 3m 16s\n- build-openstack-releasenotes http://logs.openstack.org/73/660373/7/check/build-openstack-releasenotes/423077f/html/ : SUCCESS in 4m 21s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py27-func-rabbit/8be90ad/ : SUCCESS in 5m 27s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py27-func-amqp1/b8128eb/ : FAILURE in 11m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py27-func-kafka/3e8ae4c/ : FAILURE in 4m 20s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py36-func-rabbit/2b2be88/ : SUCCESS in 5m 12s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py36-func-amqp1/4bd96b7/ : FAILURE in 12m 28s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/7/check/oslo.messaging-tox-py36-func-kafka/e749fdb/ : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-dsvm-full-rabbit/ddb0b63/ : SUCCESS in 1h 45m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/22eab4f/ : FAILURE in 31m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/4c7c213/ : FAILURE in 39m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-dsvm-full-kafka-hybrid/785b28b/ : SUCCESS in 1h 41m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-dsvm-full-kafka-centos-7/d2faec3/ : POST_FAILURE in 2h 15m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-grenade-dsvm/9611cf7/ : SUCCESS in 58m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/7/check/oslo.messaging-src-grenade-dsvm-multinode/7366eca/ : SUCCESS in 1h 21m 17s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/7/check/oslo.messaging-telemetry-dsvm-integration-rabbit/034d804/ : FAILURE in 43m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/7/check/oslo.messaging-telemetry-dsvm-integration-amqp1/fb23ac0/ : FAILURE in 37m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/7/check/oslo.messaging-telemetry-dsvm-integration-kafka/d2246b6/ : FAILURE in 48m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"df0a4485e98098be65d9a152921ecbf6d0149ed3","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-11 20:07:26.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3471bca7583218fadccf583755f605d28ebfa0b5","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-11 20:11:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b8f43e6a7d85095a0b1982fbec3a877db205cc11","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-11 20:38:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a8a23d16dd3d47995b6b1088508321c8ee08353d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-11 21:34:33.000000000","message":"Patch Set 8:\n\n(3 comments)\n\nA couple of minor things I missed the first time through. Neither are related to functional code so we could probably fix them in a followup if you\u0027d like. I do think they both need to be fixed though so we have correct docstrings and tests.","accounts_in_message":[],"_revision_number":8},{"id":"84d143d05de35efd6745e96cd1851ed8ccdb221a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:22:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/660373/8/check/tempest-full/ac41575/ : SUCCESS in 1h 46m 22s\n- tempest-full-py3 http://logs.openstack.org/73/660373/8/check/tempest-full-py3/9df6e24/ : SUCCESS in 1h 40m 15s\n- openstack-tox-cover http://logs.openstack.org/73/660373/8/check/openstack-tox-cover/838d94c/cover/ : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/8/check/openstack-tox-lower-constraints/2739106/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/8/check/openstack-tox-pep8/60d3abe/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/8/check/openstack-tox-py27/48cc439/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/8/check/openstack-tox-py36/90193e8/ : SUCCESS in 7m 31s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/8/check/openstack-tox-py37/0572dfe/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/73/660373/8/check/openstack-tox-docs/2e20090/html/ : SUCCESS in 3m 04s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py27-func-rabbit/7afb50e/ : SUCCESS in 5m 01s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py27-func-amqp1/ccc057a/ : FAILURE in 12m 27s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py27-func-kafka/efd3043/ : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py36-func-rabbit/2d5cb64/ : FAILURE in 7m 15s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py36-func-amqp1/76fca8f/ : FAILURE in 12m 27s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/8/check/oslo.messaging-tox-py36-func-kafka/ce39709/ : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-dsvm-full-rabbit/8f92051/ : SUCCESS in 1h 44m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/b5a1b1b/ : FAILURE in 29m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/ab16ce9/ : FAILURE in 40m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-dsvm-full-kafka-hybrid/26e39f8/ : SUCCESS in 1h 43m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-dsvm-full-kafka-centos-7/4c5ffc4/ : SUCCESS in 2h 02m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-grenade-dsvm/d7017b4/ : SUCCESS in 1h 06m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/8/check/oslo.messaging-src-grenade-dsvm-multinode/b1d6f1d/ : SUCCESS in 1h 19m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/8/check/oslo.messaging-telemetry-dsvm-integration-rabbit/57bde24/ : FAILURE in 44m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/8/check/oslo.messaging-telemetry-dsvm-integration-amqp1/d952965/ : FAILURE in 46m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/8/check/oslo.messaging-telemetry-dsvm-integration-kafka/1cfe940/ : FAILURE in 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cdee96dc474d245e739f5ab07d48e26bde1ff01d","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-12 08:12:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c7669fa972916773debe8df38866069c61624d43","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-12 08:21:35.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f50fac92dde380f2c71bc73d662f6d8eb24d09b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 10:41:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/9/check/tempest-full/3ee57d0/ : SUCCESS in 1h 50m 02s\n- tempest-full-py3 http://logs.openstack.org/73/660373/9/check/tempest-full-py3/c5e185b/ : SUCCESS in 1h 29m 58s\n- openstack-tox-cover http://logs.openstack.org/73/660373/9/check/openstack-tox-cover/a7dc330/cover/ : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/9/check/openstack-tox-lower-constraints/962e3af/ : SUCCESS in 6m 10s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/9/check/openstack-tox-pep8/241c1ee/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/9/check/openstack-tox-py27/461ee8e/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/9/check/openstack-tox-py36/9b021d5/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/9/check/openstack-tox-py37/bf275e0/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/73/660373/9/check/openstack-tox-docs/2fcfed0/html/ : SUCCESS in 3m 17s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py27-func-rabbit/c51ef34/ : FAILURE in 6m 29s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py27-func-amqp1/2fe30e0/ : FAILURE in 11m 26s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py27-func-kafka/2856b5e/ : FAILURE in 4m 01s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py36-func-rabbit/2dc644b/ : SUCCESS in 5m 19s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py36-func-amqp1/b1ba6ca/ : FAILURE in 12m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/9/check/oslo.messaging-tox-py36-func-kafka/552fc9d/ : FAILURE in 4m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-dsvm-full-rabbit/d0fa7e1/ : SUCCESS in 2h 15m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/92eb455/ : FAILURE in 31m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/b2ec73e/ : FAILURE in 36m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-dsvm-full-kafka-hybrid/04aef3b/ : SUCCESS in 1h 46m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-dsvm-full-kafka-centos-7/2557e9c/ : SUCCESS in 1h 59m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-grenade-dsvm/f1fc22b/ : SUCCESS in 59m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/9/check/oslo.messaging-src-grenade-dsvm-multinode/f4eaba9/ : SUCCESS in 1h 09m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/9/check/oslo.messaging-telemetry-dsvm-integration-rabbit/075d3aa/ : FAILURE in 46m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/9/check/oslo.messaging-telemetry-dsvm-integration-amqp1/521c04d/ : FAILURE in 39m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/9/check/oslo.messaging-telemetry-dsvm-integration-kafka/4661e09/ : FAILURE in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2aa843094aa2aa0ea489a176806f7aee862ec2b0","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-12 10:50:20.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"da4077495770089e9c3f08ac28bd938f0f29e348","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-12 12:38:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"32147c773b818c93f37882742d8e44d144eb03c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-12 12:38:59.000000000","message":"Patch Set 9: Code-Review-1","accounts_in_message":[],"_revision_number":9},{"id":"66ad4ff4e6ff7951135df9728d71be40139539c3","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-12 12:43:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"357f28aae357389aeb9db5d18d1a2270477362d8","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-12 12:44:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"78be8743a2652ebc38208db7fcc9f3a048e8e337","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-12 12:52:28.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7f52e0cec1e4eff063bcf354d6dbc6d43afcce5d","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-06-12 14:06:14.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nI like this addition to the API, however I do have two concerns I\u0027d like to see addressed - more about policy than the code itself:\n\nThe term \u0027options\u0027 is pretty generic and I think it can lead to confusion when introducing new parameters to the send API.  Specifically when do I add something to \"options\" vs when a new keyword parameter is warranted?  Why isn\u0027t \"retry\" an option and \u0027mandatory\u0027 is? etc.\n\nI think we need to clarify what we consider \"options\" and what we consider \"things added to the parameter list\".\n\nFor example - not necessarily a good one, but an example:\n\n1) Generic RPC concepts - timeout, retry (hmm), etc, are keyword parameters\n2) we use the keyword \"delivery_options\" or \"delivery_flags\" (blah) that affect the quality of service for the message delivery itself (\u0027confirm-receipt\u0027, \u0027at-most-once\u0027, \u0027at-least-once\u0027, etc).\n\nWhich brings me to my second concern - what if a transport does NOT support a given option?  For example the term \u0027mandatory\u0027 is rabbit specific - the AMQP ISO standard does not implement a \u0027mandatory\u0027 flag.  And I don\u0027t know enough about Kafka\u0027s protocol either way.","accounts_in_message":[],"_revision_number":10},{"id":"3687c59e2e724b7bffd90e66249382356931c4b3","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-06-12 14:30:08.000000000","message":"Patch Set 10:\n\nSorry for the comment bombing :(\n\nI wanted to point out that it was my suggestion to Gabriele to add an options parameter to send in order to make the \u0027mandatory\u0027 flag optional.\n\nMy concern was that if we simply turned on \u0027mandatory\u0027 then we\u0027ve introduced a new behavior and that things would break.\n\nHowever after reading Sean\u0027s comments I may have been wrong with that approach.\n\nI think it can be argued that - in some cases - use of \u0027mandatory\u0027 should be, well, _mandatory_.  Here\u0027s my thoughts:\n\no.m. supports 3 flavors of RPC calls:\n\na \u0027call\u0027 - blocks for a reply\na \u0027cast\u0027 - simple send no reply expected\na \u0027fanout cast\u0027 - a cast, but the message is broadcast to all subscribers\n\nI think that \u0027mandatory\u0027 should be turned on by default for the \u0027call\u0027 case.  It should be an option (default off) in the cast/fanout case.\n\nRight now IIRC when o.m. does a \u0027call\u0027 it sends the request message then blocks for the reply until timeout even if there is no consumer for the request.  What \u0027mandatory\u0027 would do in this case is to fail immediately if there is no consumer rather than time out (Gabriele correct me if I\u0027m wrong).  \n\nThat\u0027s a good thing, right?  In fact I\u0027d say that it is a bug _not_ to use the mandatory flag in the \u0027call\u0027 case.\n\nOptions?","accounts_in_message":[],"_revision_number":10},{"id":"d3ed7c375bc715124ad0f68de1a5a09a54725164","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:02:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/660373/10/check/tempest-full/888d231/ : SUCCESS in 1h 58m 41s\n- tempest-full-py3 http://logs.openstack.org/73/660373/10/check/tempest-full-py3/eb2a43d/ : SUCCESS in 1h 33m 25s\n- openstack-tox-cover http://logs.openstack.org/73/660373/10/check/openstack-tox-cover/40d36a1/cover/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/10/check/openstack-tox-lower-constraints/df86b3d/ : SUCCESS in 5m 30s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/10/check/openstack-tox-pep8/89442ba/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/10/check/openstack-tox-py27/b90a052/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/10/check/openstack-tox-py36/63825af/ : SUCCESS in 6m 20s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/10/check/openstack-tox-py37/d17a54a/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/73/660373/10/check/openstack-tox-docs/31d0d1b/html/ : SUCCESS in 2m 47s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py27-func-rabbit/ba2aeef/ : SUCCESS in 5m 34s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py27-func-amqp1/7d2c41c/ : FAILURE in 11m 37s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py27-func-kafka/d36d9b9/ : FAILURE in 8m 43s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py36-func-rabbit/7b509a0/ : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py36-func-amqp1/fdc931c/ : FAILURE in 11m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/10/check/oslo.messaging-tox-py36-func-kafka/8355275/ : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-dsvm-full-rabbit/b8358e1/ : SUCCESS in 2h 11m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/81bc5f4/ : FAILURE in 30m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/a84de5d/ : FAILURE in 33m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3355215/ : SUCCESS in 2h 04m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-dsvm-full-kafka-centos-7/2d7a175/ : SUCCESS in 2h 09m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-grenade-dsvm/55dbe91/ : SUCCESS in 1h 00m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/10/check/oslo.messaging-src-grenade-dsvm-multinode/9687c46/ : SUCCESS in 1h 11m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/10/check/oslo.messaging-telemetry-dsvm-integration-rabbit/c23d42f/ : FAILURE in 45m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/10/check/oslo.messaging-telemetry-dsvm-integration-amqp1/695309d/ : FAILURE in 34m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/10/check/oslo.messaging-telemetry-dsvm-integration-kafka/94b9cc3/ : FAILURE in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"643121ecccf014c566f9619add40aec5aa2c31e1","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 08:31:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3fa88bdb827bf365b76b43a6cb1dcfc411c509b1","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 08:37:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"8078bfcbb06ad373035bd334033efee0e8b9a58e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:10:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/12/check/tempest-full/0ac2973/ : FAILURE in 1h 47m 18s\n- tempest-full-py3 http://logs.openstack.org/73/660373/12/check/tempest-full-py3/58a727b/ : SUCCESS in 1h 48m 07s\n- openstack-tox-cover http://logs.openstack.org/73/660373/12/check/openstack-tox-cover/42cb35f/cover/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/12/check/openstack-tox-lower-constraints/54ee9a3/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/12/check/openstack-tox-pep8/b6d3d3e/ : SUCCESS in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/12/check/openstack-tox-py27/831e7e6/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/12/check/openstack-tox-py36/fd1b4fe/ : SUCCESS in 5m 48s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/12/check/openstack-tox-py37/31d4d26/ : SUCCESS in 6m 59s\n- openstack-tox-docs http://logs.openstack.org/73/660373/12/check/openstack-tox-docs/23329b8/html/ : SUCCESS in 3m 04s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py27-func-rabbit/6a14a34/ : SUCCESS in 6m 53s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py27-func-amqp1/a1b669c/ : FAILURE in 12m 59s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py27-func-kafka/3b7fbc5/ : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py36-func-rabbit/f0c6277/ : SUCCESS in 5m 14s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py36-func-amqp1/e788dca/ : FAILURE in 13m 21s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/12/check/oslo.messaging-tox-py36-func-kafka/50cb42c/ : FAILURE in 4m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-dsvm-full-rabbit/6c76618/ : SUCCESS in 1h 38m 33s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/c942326/ : FAILURE in 31m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/1a38f61/ : FAILURE in 43m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-dsvm-full-kafka-hybrid/1b0fa67/ : SUCCESS in 1h 52m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-dsvm-full-kafka-centos-7/a20b487/ : SUCCESS in 2h 25m 26s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-grenade-dsvm/3abcd4d/ : SUCCESS in 1h 04m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/12/check/oslo.messaging-src-grenade-dsvm-multinode/a96281a/ : SUCCESS in 1h 04m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/12/check/oslo.messaging-telemetry-dsvm-integration-rabbit/062d165/ : FAILURE in 48m 13s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/12/check/oslo.messaging-telemetry-dsvm-integration-amqp1/e2b1307/ : FAILURE in 49m 46s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/12/check/oslo.messaging-telemetry-dsvm-integration-kafka/41cf60a/ : FAILURE in 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6834788c165288ab9b6d35eafc949e11551e343d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 11:19:26.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"43cef1b7e18d65796c62b30e4624006f988c2c0c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 11:31:03.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)\n\n\u003e Sorry for the comment bombing :(\n \u003e \n \u003e I wanted to point out that it was my suggestion to Gabriele to add\n \u003e an options parameter to send in order to make the \u0027mandatory\u0027 flag\n \u003e optional.\n \u003e \n \u003e My concern was that if we simply turned on \u0027mandatory\u0027 then we\u0027ve\n \u003e introduced a new behavior and that things would break.\n \u003e \n \u003e However after reading Sean\u0027s comments I may have been wrong with\n \u003e that approach.\n \u003e \n \u003e I think it can be argued that - in some cases - use of \u0027mandatory\u0027\n \u003e should be, well, _mandatory_.  Here\u0027s my thoughts:\n \u003e \n \u003e o.m. supports 3 flavors of RPC calls:\n \u003e \n \u003e a \u0027call\u0027 - blocks for a reply\n \u003e a \u0027cast\u0027 - simple send no reply expected\n \u003e a \u0027fanout cast\u0027 - a cast, but the message is broadcast to all\n \u003e subscribers\n \u003e \n \u003e I think that \u0027mandatory\u0027 should be turned on by default for the\n \u003e \u0027call\u0027 case.  It should be an option (default off) in the\n \u003e cast/fanout case.\n\nNot sure to see all the possible use cases and issues here, but I think it make sense.\n\n \u003e Right now IIRC when o.m. does a \u0027call\u0027 it sends the request message\n \u003e then blocks for the reply until timeout even if there is no\n \u003e consumer for the request.  What \u0027mandatory\u0027 would do in this case\n \u003e is to fail immediately if there is no consumer rather than time out\n \u003e (Gabriele correct me if I\u0027m wrong).\n \u003e \n \u003e That\u0027s a good thing, right?  In fact I\u0027d say that it is a bug _not_\n \u003e to use the mandatory flag in the \u0027call\u0027 case.\n \u003e \n \u003e Options?","accounts_in_message":[],"_revision_number":12},{"id":"efa6954dc2adfbe30dce9d1370f44fc0e0a7f70f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 11:31:51.000000000","message":"Patch Set 12: -Code-Review","accounts_in_message":[],"_revision_number":12},{"id":"10cae1455c93cfbc4ed2a81de58479ed4cddb52e","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 11:40:22.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"020da6e027ad5a30425cabbde3e2bad3c8d03066","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 11:44:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7ba5a8a30dab9dedfc06345610234d210cb347aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 11:47:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"44851e42cfe22ed7e5be8d90dfb3b67e5199c7b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 11:49:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2ce8c812da7f6f51ad0ef7f45c97145a065a6607","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 12:16:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5e27047d17ae07f717644b8330be9f1c0ca36ecc","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-14 13:49:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cee8a8e90c4fa21b3d6e175884c217f0a16a8756","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-14 13:57:23.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)\n\nWaiting for the CI and kguisti recheck","accounts_in_message":[],"_revision_number":13},{"id":"2f5a035f6646793ffbfd4be35917f4beabe45ed1","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-06-14 16:19:50.000000000","message":"Patch Set 13: Code-Review+1\n\nWill +2 when zuul is pleased.\n\nThanks Gabriele","accounts_in_message":[],"_revision_number":13},{"id":"c4ab45b2a35341c450b0ab2e9b1c28d915ad9692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 18:05:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/660373/13/check/tempest-full/87c0329/ : SUCCESS in 1h 44m 11s\n- tempest-full-py3 http://logs.openstack.org/73/660373/13/check/tempest-full-py3/3ebc12e/ : SUCCESS in 1h 44m 25s\n- openstack-tox-cover http://logs.openstack.org/73/660373/13/check/openstack-tox-cover/97dae69/cover/ : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/13/check/openstack-tox-lower-constraints/8e5ae6d/ : SUCCESS in 8m 36s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/13/check/openstack-tox-pep8/170d481/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/13/check/openstack-tox-py27/6dccf74/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/13/check/openstack-tox-py36/9f76d0d/ : SUCCESS in 5m 14s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/13/check/openstack-tox-py37/78f6f65/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/73/660373/13/check/openstack-tox-docs/2de28b4/html/ : SUCCESS in 3m 37s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-rabbit/95a5d39/ : SUCCESS in 5m 05s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-amqp1/c7f67e2/ : FAILURE in 14m 05s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-kafka/031af45/ : FAILURE in 3m 43s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-rabbit/d3ab8a0/ : SUCCESS in 5m 53s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-amqp1/332c3c0/ : FAILURE in 12m 00s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-kafka/483cdd9/ : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-rabbit/6743800/ : SUCCESS in 1h 40m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/ea97cd2/ : FAILURE in 25m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/96bec7a/ : FAILURE in 39m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-kafka-hybrid/d258520/ : SUCCESS in 1h 49m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-kafka-centos-7/fe30e38/ : SUCCESS in 2h 17m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-grenade-dsvm/85f264b/ : SUCCESS in 1h 10m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-grenade-dsvm-multinode/d156ded/ : SUCCESS in 1h 07m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-rabbit/b1866a6/ : FAILURE in 47m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-amqp1/c1d7752/ : FAILURE in 1h 07m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-kafka/93594e9/ : FAILURE in 1h 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f55c4d20a91493895f5af4bb6e891e2313ce5716","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-06-15 12:04:04.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"d4608856bd0a92c121141e41df380631f34352a7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-17 07:47:41.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"f942ddc0234a2b711f423163671a6a0900d3375a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 07:47:53.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"ed138850c8dfaa41d7aa62d85d09d43f0a1495a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 10:03:32.000000000","message":"Patch Set 13: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/73/660373/13/gate/tempest-full/521b15d/ : SUCCESS in 1h 44m 14s\n- tempest-full-py3 http://logs.openstack.org/73/660373/13/gate/tempest-full-py3/623cb52/ : SUCCESS in 1h 54m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/13/gate/openstack-tox-lower-constraints/9c8e0e8/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/13/gate/openstack-tox-pep8/5ceaa21/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py27/ec4803e/ : SUCCESS in 7m 15s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py36/fa50c74/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py37/e616cf2/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/73/660373/13/gate/openstack-tox-docs/e6c9788/html/ : SUCCESS in 4m 19s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/13/gate/oslo.messaging-tox-py27-func-rabbit/1725ec7/ : SUCCESS in 5m 42s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/13/gate/oslo.messaging-src-dsvm-full-rabbit/1608b34/ : FAILURE in 1h 45m 15s","accounts_in_message":[],"_revision_number":13},{"id":"5224978914189a721b9838b19d68acb546c5072e","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-17 12:27:31.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"8f2c2a70ca538366f825cf7a20e467215a5bcb4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:39:47.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/660373/13/check/tempest-full/acea53e/ : SUCCESS in 1h 44m 32s\n- tempest-full-py3 http://logs.openstack.org/73/660373/13/check/tempest-full-py3/bedb093/ : SUCCESS in 1h 30m 05s\n- openstack-tox-cover http://logs.openstack.org/73/660373/13/check/openstack-tox-cover/b410920/cover/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/13/check/openstack-tox-lower-constraints/dd8aac4/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/13/check/openstack-tox-pep8/b3951eb/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/13/check/openstack-tox-py27/c6076ea/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/13/check/openstack-tox-py36/0738485/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/13/check/openstack-tox-py37/aff7b30/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/73/660373/13/check/openstack-tox-docs/a770512/html/ : SUCCESS in 2m 56s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-rabbit/f011147/ : SUCCESS in 5m 20s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-amqp1/a74caeb/ : FAILURE in 11m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py27-func-kafka/ab01c3c/ : FAILURE in 11m 18s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-rabbit/d050453/ : SUCCESS in 5m 51s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-amqp1/9f0c46a/ : FAILURE in 11m 55s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-tox-py36-func-kafka/77d213a/ : FAILURE in 4m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-rabbit/f708b4d/ : SUCCESS in 1h 28m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/8b042fe/ : FAILURE in 36m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/94294f7/ : FAILURE in 41m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-kafka-hybrid/37a3c1e/ : SUCCESS in 1h 50m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-dsvm-full-kafka-centos-7/845ab03/ : SUCCESS in 2h 05m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-grenade-dsvm/25092ec/ : SUCCESS in 1h 10m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/660373/13/check/oslo.messaging-src-grenade-dsvm-multinode/0c1d324/ : SUCCESS in 1h 10m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-rabbit/7847c5a/ : FAILURE in 37m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-amqp1/4c98574/ : FAILURE in 46m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/660373/13/check/oslo.messaging-telemetry-dsvm-integration-kafka/8ee0799/ : FAILURE in 53m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f052013ce0d91c2d5f5129f879b8aaad4c142fff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:39:59.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"0f07a0a4987030b64f791abb49818b71e31e3325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:35:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"7895506fe18ad112dfe60872d533bdc424406d07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:35:55.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/73/660373/13/gate/tempest-full/5d4c7e3/ : SUCCESS in 1h 33m 05s\n- tempest-full-py3 http://logs.openstack.org/73/660373/13/gate/tempest-full-py3/740ce76/ : SUCCESS in 1h 41m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/660373/13/gate/openstack-tox-lower-constraints/2f35993/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/73/660373/13/gate/openstack-tox-pep8/0758a35/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py27/3ca14eb/ : SUCCESS in 5m 04s\n- openstack-tox-py36 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py36/85e996f/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/73/660373/13/gate/openstack-tox-py37/0a956bc/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/73/660373/13/gate/openstack-tox-docs/019858a/html/ : SUCCESS in 4m 16s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/660373/13/gate/oslo.messaging-tox-py27-func-rabbit/30f37e8/ : SUCCESS in 5m 28s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/660373/13/gate/oslo.messaging-src-dsvm-full-rabbit/228e969/ : SUCCESS in 1h 52m 21s","accounts_in_message":[],"_revision_number":13},{"id":"507b4e86d7430a381b28b71785f03c29ccc67a52","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-10-17 17:39:58.000000000","message":"Patch Set 13:\n\nJust wanted to link a couple of references of the IRC discussion that motivated interest in this change from the nova side, for those reading up on this change and associated bug in the future:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-06-06.log.html#t2019-06-06T22:47:09\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-06-07.log.html#t2019-06-07T01:23:40","accounts_in_message":[],"_revision_number":13},{"id":"25e76baf178a734beb44e82a3952b0b9c3d2eef7","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-11 00:29:26.000000000","message":"Patch Set 13: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 9931198ab4b9b80fd76eb619c4dbda9fd96615a7","accounts_in_message":[],"_revision_number":13},{"id":"e26d4c118b7d5ed0574f262d8798cb59e9b9d9c5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-05-12 06:08:35.000000000","message":"Patch Set 13: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ab4f87aca55e3ce84a2b90f7229c212853193981","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"83266cc6ea7dbe682f8fdb960a548c68fd1235b5","revisions":{"dbc20fa420a6a11fda63d272a12848926c4f40fb":{"kind":"REWORK","_number":1,"created":"2019-05-21 12:42:10.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/1"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"subject":"WIP - Add the mandatory flag","message":"WIP - Add the mandatory flag\n\nAdd mandatory flag to handle the no-routed messages.\nIn this way it is possible to know if a message reaches a\nqueue or not.\n\nIt raises an exception in case a message is not routed.\n\nNote:\n  - This is a WIP\n  - This is just a an idea, I wanted to add a\n    generic parameter called \"option\",\n    so we can add more options during the publish.\n\nSee the gist[1] for details.\n\nCloses-bug: #1661510\n\n1 - https://gist.github.com/Gsantomaggio/bff62fe28210622cfa9d32d164412800\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dbc20fa420a6a11fda63d272a12848926c4f40fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dbc20fa420a6a11fda63d272a12848926c4f40fb"}]},"branch":"refs/heads/master"},"e27e136e1d181c059972a9e875d3bff99aa2cd4f":{"kind":"REWORK","_number":2,"created":"2019-06-03 09:33:56.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/2"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-03 09:31:31.000000000","tz":120},"subject":"WIP - Add the mandatory flag","message":"WIP - Add the mandatory flag\n\nAdd mandatory flag to handle the no-routed messages.\nIn this way it is possible to know if a message reaches a\nqueue or not.\n\nIt raises an exception in case a message is not routed.\n\nNote:\n  - This is a WIP\n  - This is just a an idea, I wanted to add a\n    generic parameter called \"option\",\n    so we can add more options during the publish.\n\nSee the repo[1] for details.\n\nCloses-bug: #1661510\n\n1 - https://github.com/Gsantomaggio/rabbitmq-utils/\n    tree/master/openstack/mandatory_test\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e27e136e1d181c059972a9e875d3bff99aa2cd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e27e136e1d181c059972a9e875d3bff99aa2cd4f"}]},"branch":"refs/heads/master"},"6744c9e8bc906619052e9f50b008ec2a235766ce":{"kind":"REWORK","_number":3,"created":"2019-06-10 11:10:01.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/3"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-10 11:09:16.000000000","tz":120},"subject":"WIP - Add the mandatory flag","message":"WIP - Add the mandatory flag\n\nAdd mandatory flag to handle the no-routed messages.\nIn this way it is possible to know if a message reaches a\nqueue or not.\n\nIt raises an exception in case a message is not routed.\n\nNote:\n  - This is a WIP\n  - I wanted to add a\n    generic parameter called \"option\",\n    so we can add more options during the publish.\n\nSee the repo[1] for details.\n\nCloses-bug: #1661510\n\n1 - https://github.com/Gsantomaggio/rabbitmq-utils/\n    tree/master/openstack/mandatory_test\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6744c9e8bc906619052e9f50b008ec2a235766ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6744c9e8bc906619052e9f50b008ec2a235766ce"}]},"branch":"refs/heads/master"},"9b0b4319215f5e87b3dcbc798e0866d68223448e":{"kind":"REWORK","_number":4,"created":"2019-06-10 16:22:03.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/4"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-10 16:18:43.000000000","tz":120},"subject":"WIP - Add the mandatory flag","message":"WIP - Add the mandatory flag\n\nAdd mandatory flag to handle the no-routed messages.\nIn this way it is possible to know if a message reaches a\nqueue or not.\n\nIt raises an exception in case a message is not routed.\n\nNote:\n  - This is a WIP\n  - Needs Kombu 4.1.6, because of [2]\n  - I wanted to add a\n    generic parameter called \"option\",\n    so we can add more options during the publish.\n\nSee the repo[1] for details.\n\nCloses-bug: #1661510\n\n1- https://github.com/Gsantomaggio/rabbitmq-utils/\n    tree/master/openstack/mandatory_test\n2- https://github.com/celery/kombu/pull/1053\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b0b4319215f5e87b3dcbc798e0866d68223448e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b0b4319215f5e87b3dcbc798e0866d68223448e"}]},"branch":"refs/heads/master"},"9e2ed7512a07af1a7801ca83c8023120457ceae3":{"kind":"REWORK","_number":5,"created":"2019-06-11 14:58:39.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/5"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-11 14:54:57.000000000","tz":120},"subject":"WIP - Add options parameter","message":"WIP - Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in rabbitmq can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doorsn for new implementations\n\nPart of: #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9e2ed7512a07af1a7801ca83c8023120457ceae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9e2ed7512a07af1a7801ca83c8023120457ceae3"}]},"branch":"refs/heads/master"},"df775ab6e425c969fcadf2c09732da4ddb940c89":{"kind":"REWORK","_number":6,"created":"2019-06-11 15:04:32.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/6"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-11 15:04:08.000000000","tz":120},"subject":"Add options parameter","message":"Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in rabbitmq can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doorsn for new implementations\n\nPart of: #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/df775ab6e425c969fcadf2c09732da4ddb940c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/df775ab6e425c969fcadf2c09732da4ddb940c89"}]},"branch":"refs/heads/master"},"f083135369850c31a87e0b43c73f132a37d287f2":{"kind":"REWORK","_number":7,"created":"2019-06-11 15:15:32.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/7"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-11 15:15:07.000000000","tz":120},"subject":"Add options parameter","message":"Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in rabbitmq can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doorsn for new implementations\n\nThis part of: #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f083135369850c31a87e0b43c73f132a37d287f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f083135369850c31a87e0b43c73f132a37d287f2"}]},"branch":"refs/heads/master"},"50a59fad26de0a8a2bb1c4ed31211442cc4c25f6":{"kind":"REWORK","_number":8,"created":"2019-06-11 20:11:49.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/8"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-11 20:11:21.000000000","tz":120},"subject":"Add options parameter","message":"Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in RabbitMQ can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doorsn for new implementations\n\nThis is part of: #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/50a59fad26de0a8a2bb1c4ed31211442cc4c25f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/50a59fad26de0a8a2bb1c4ed31211442cc4c25f6"}]},"branch":"refs/heads/master"},"082e2a0ffd6c7c63d9334cc079bd6731cd7f91ef":{"kind":"REWORK","_number":9,"created":"2019-06-12 08:12:27.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/9"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-12 08:11:28.000000000","tz":120},"subject":"Add options parameter","message":"Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in RabbitMQ can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doors for new implementations\n\nThis is part of: #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/082e2a0ffd6c7c63d9334cc079bd6731cd7f91ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/082e2a0ffd6c7c63d9334cc079bd6731cd7f91ef"}]},"branch":"refs/heads/master"},"5cbc34e946b45d730c2e5664de2e9c78231bf441":{"kind":"REWORK","_number":10,"created":"2019-06-12 12:43:01.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/10"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-12 12:41:43.000000000","tz":120},"subject":"Add options parameter","message":"Add options parameter\n\nWith this new parameter is possible to pass other values\nto the send function.\nFor example in RabbitMQ can be used for the mandatory flag.\n\nNote:\n  - The options parameter is not actually used.\n  - This new feature opens the doors for new implementations\n\nThis is part of:  #1661510\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cbc34e946b45d730c2e5664de2e9c78231bf441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cbc34e946b45d730c2e5664de2e9c78231bf441"}]},"branch":"refs/heads/master"},"43d16dce3a13a83394ad2f15f3c59adcfd9b9bec":{"kind":"REWORK","_number":11,"created":"2019-06-14 08:31:15.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/11"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-14 08:30:47.000000000","tz":120},"subject":"Add transport_options parameter","message":"Add transport_options parameter\n\nWith this new parameter is possible to pass other parameters\nfrom the client to the drivers.\nSo it is possible to tune the driver behavior.\n\nFor example can be used to send the mandatory flag in RabbitMQ\n\nNote:\n  - The transport_options parameter is not actually used (yet).\n  - This part of blueprint transport-options  (first part)\n\nImplements blueprint: transport-options\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43d16dce3a13a83394ad2f15f3c59adcfd9b9bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43d16dce3a13a83394ad2f15f3c59adcfd9b9bec"}]},"branch":"refs/heads/master"},"e561e8c4a3016bef148db586859de366c7a54789":{"kind":"REWORK","_number":12,"created":"2019-06-14 08:37:21.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/12"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-14 08:36:31.000000000","tz":120},"subject":"Add transport_options parameter","message":"Add transport_options parameter\n\nWith this new parameter is possible to pass other parameters\nfrom the client to the drivers.\nSo it is possible to tune the driver behavior.\n\nFor example can be used to send the mandatory flag in RabbitMQ\n\nNote:\n  - The transport_options parameter is not actually used (yet).\n  - This part of blueprint transport-options  (first part)\n\nImplements: blueprint transport-options\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e561e8c4a3016bef148db586859de366c7a54789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e561e8c4a3016bef148db586859de366c7a54789"}]},"branch":"refs/heads/master"},"83266cc6ea7dbe682f8fdb960a548c68fd1235b5":{"kind":"REWORK","_number":13,"created":"2019-06-14 13:49:36.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/660373/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/660373/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/660373/13"}}},"commit":{"parents":[{"commit":"40c25c2bde6d2f5a756e7169060b7ce389caf174","subject":"Merge \"Remove log translation and i18n\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40c25c2bde6d2f5a756e7169060b7ce389caf174"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-05-21 12:39:55.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-14 13:47:39.000000000","tz":120},"subject":"Add transport_options parameter","message":"Add transport_options parameter\n\nWith this new parameter is possible to pass other parameters\nfrom the client to the drivers.\nSo it is possible to tune the driver behavior.\n\nFor example can be used to send the mandatory flag in RabbitMQ\n\nNote:\n  - The transport_options parameter is not actually used (yet).\n  - This part of blueprint transport-options  (first part)\n\nImplements: blueprint transport-options\nThe blueprint link is\nhttps://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n\nChange-Id: Iff23a9344c2c87259cf36b0d14c0a9fc075a2a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/83266cc6ea7dbe682f8fdb960a548c68fd1235b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/83266cc6ea7dbe682f8fdb960a548c68fd1235b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
