)]}'
{"id":"openstack%2Foslo.messaging~660389","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~Ie6b379f2c99862c37891ac03c52464e07bc2b2cc","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"bug/1828841","hashtags":[],"change_id":"Ie6b379f2c99862c37891ac03c52464e07bc2b2cc","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-21 13:31:06.000000000","updated":"2019-05-24 15:47:44.000000000","submitted":"2019-05-24 15:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660389-1558712864448-66e7a990","meta_rev_id":"9ad2f4b8b008c7e9be05b8294dde42da97da36cd","_number":660389,"virtual_id_number":660389,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-05-24 15:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-24 14:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-22 12:47:55.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-21 14:05:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-24 14:05:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 14:05:42.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-22 12:47:55.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-24 14:05:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-24 15:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88141e7e3cef6ee87199c834461381be581bff89","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 13:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc2b8a716f745ce48f75edf70d9847d9e0a6e21c","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-21 14:05:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1312fc14233085fa91610700d0a3b02f92fe50e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:44:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/89/660389/1/check/requirements-check/b93026d/ : SUCCESS in 3m 12s\n- tempest-full http://logs.openstack.org/89/660389/1/check/tempest-full/025b838/ : SUCCESS in 1h 46m 39s\n- tempest-full-py3 http://logs.openstack.org/89/660389/1/check/tempest-full-py3/40b2ade/ : SUCCESS in 1h 32m 59s\n- openstack-tox-cover http://logs.openstack.org/89/660389/1/check/openstack-tox-cover/b6077ff/cover/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/660389/1/check/openstack-tox-lower-constraints/520a122/ : SUCCESS in 4m 23s\n- openstack-tox-pep8 http://logs.openstack.org/89/660389/1/check/openstack-tox-pep8/42a1f59/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/89/660389/1/check/openstack-tox-py27/9854e72/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/89/660389/1/check/openstack-tox-py36/829a969/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/89/660389/1/check/openstack-tox-py37/1d40be0/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/89/660389/1/check/openstack-tox-docs/5183753/html/ : SUCCESS in 3m 12s\n- build-openstack-releasenotes http://logs.openstack.org/89/660389/1/check/build-openstack-releasenotes/2af91f3/html/ : SUCCESS in 2m 51s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py27-func-rabbit/edfc1ff/ : SUCCESS in 6m 53s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py27-func-amqp1/fcd2e5d/ : SUCCESS in 4m 06s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py27-func-kafka/0dd022f/ : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py35-func-rabbit/aeb11cd/ : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py35-func-amqp1/dd7b525/ : SUCCESS in 4m 05s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/89/660389/1/check/oslo.messaging-tox-py35-func-kafka/28786b6/ : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-dsvm-full-rabbit/8a47c9b/ : SUCCESS in 1h 56m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/9586f1d/ : SUCCESS in 1h 57m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/14b8b7d/ : FAILURE in 1h 30m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/32c830f/ : FAILURE in 19m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/110e04e/ : FAILURE in 20m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-grenade-dsvm/5e639b3/ : SUCCESS in 54m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/89/660389/1/check/oslo.messaging-src-grenade-dsvm-multinode/17ef310/ : SUCCESS in 59m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/89/660389/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/c8c1f31/ : FAILURE in 47m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/89/660389/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/4602a18/ : FAILURE in 42m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/89/660389/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/82a22eb/ : FAILURE in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0339020a6c681dbafbd2b4c4acdc8c7beccbcec1","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-22 12:47:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"410130f8abbded57d38dfa7c628d81ce73c214b8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-24 14:05:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMerged to master and appropriate to backport. I believe the newest release of bandit fixes the issues this is avoiding, so could potentially change to only exclude 1.6.0, but that is something that can be addressed on master later.","accounts_in_message":[],"_revision_number":1},{"id":"cb9d3d3630655a16bd0c1b6cf0929f5e6ba60f34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 14:05:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28dcd7b29ea954baa262447e6bdc96d0b3e4cdcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:47:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6676cd65d3e24b7988b0edcf8a962e24161dc37d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 15:47:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/89/660389/1/gate/requirements-check/2c155c1/ : SUCCESS in 3m 20s\n- tempest-full http://logs.openstack.org/89/660389/1/gate/tempest-full/11de906/ : SUCCESS in 1h 40m 40s\n- tempest-full-py3 http://logs.openstack.org/89/660389/1/gate/tempest-full-py3/0d5a4f2/ : SUCCESS in 1h 27m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/660389/1/gate/openstack-tox-lower-constraints/7f721ad/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/89/660389/1/gate/openstack-tox-pep8/e9f9f3f/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/89/660389/1/gate/openstack-tox-py27/ab4a440/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/89/660389/1/gate/openstack-tox-py36/8ca3cb3/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/89/660389/1/gate/openstack-tox-py37/7bb3505/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/89/660389/1/gate/openstack-tox-docs/a33e717/html/ : SUCCESS in 2m 26s\n- build-openstack-releasenotes http://logs.openstack.org/89/660389/1/gate/build-openstack-releasenotes/2d0e7db/html/ : SUCCESS in 3m 38s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/89/660389/1/gate/oslo.messaging-tox-py27-func-rabbit/0436618/ : SUCCESS in 4m 46s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/89/660389/1/gate/oslo.messaging-src-dsvm-full-rabbit/58c7e2a/ : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f91b0a2115f3c0a817abb624d65f8f0dcad40b2","revisions":{"0f91b0a2115f3c0a817abb624d65f8f0dcad40b2":{"kind":"REWORK","_number":1,"created":"2019-05-21 13:31:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/89/660389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/660389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/660389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/660389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/660389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/660389/1"}}},"commit":{"parents":[{"commit":"f276f0853afcda949a665d2fa12d712e96d7115f","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f276f0853afcda949a665d2fa12d712e96d7115f"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-05-13 14:55:23.000000000","tz":-240},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-21 13:24:23.000000000","tz":120},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start\ncapping it there as well.\n\nChange-Id: Ie6b379f2c99862c37891ac03c52464e07bc2b2cc\n(cherry picked from commit 4f467fbd73092aadda8b42e517bb60a4782064ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f91b0a2115f3c0a817abb624d65f8f0dcad40b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f91b0a2115f3c0a817abb624d65f8f0dcad40b2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
