)]}'
{"id":"openstack%2Foslo.messaging~660390","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~Ie6b379f2c99862c37891ac03c52464e07bc2b2cc","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bug/1828841","hashtags":[],"change_id":"Ie6b379f2c99862c37891ac03c52464e07bc2b2cc","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-21 13:32:54.000000000","updated":"2019-06-06 11:03:13.000000000","submitted":"2019-06-06 11:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660390-1559818993047-d3725ed7","meta_rev_id":"b6dc3c7dbe2bfec348abf9d99ffb7ba93b70ab6c","_number":660390,"virtual_id_number":660390,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-06 11:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-05-22 12:49:00.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-21 14:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-06-06 09:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-06-06 09:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 14:07:13.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-05-22 12:49:00.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-06-06 09:09:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-06 11:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26e987e174d5a1b4c65c9c0c5585a7502405035c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-21 13:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1364666c1a5be2514a82476e74c4d60d9f3be13","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-21 14:07:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c1270f063abc9198a860bb3c7d88300d8164fe1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/660390/1/check/openstack-tox-pep8/c549205/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/90/660390/1/check/openstack-tox-py27/582fab3/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/90/660390/1/check/openstack-tox-py35/25d39ab/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/90/660390/1/check/openstack-tox-docs/47a5dea/html/ : SUCCESS in 3m 07s\n- requirements-check http://logs.openstack.org/90/660390/1/check/requirements-check/6cf5c30/ : SUCCESS in 2m 40s\n- tempest-full http://logs.openstack.org/90/660390/1/check/tempest-full/67f2e92/ : SUCCESS in 1h 26m 30s\n- build-openstack-releasenotes http://logs.openstack.org/90/660390/1/check/build-openstack-releasenotes/b491f14/html/ : SUCCESS in 3m 33s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py27-func-rabbit/cc8182c/ : SUCCESS in 5m 18s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py27-func-amqp1/999732b/ : SUCCESS in 3m 43s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py27-func-kafka/578bb6d/ : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py35-func-rabbit/e131091/ : SUCCESS in 4m 54s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py35-func-amqp1/8ecc5f8/ : SUCCESS in 4m 29s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/90/660390/1/check/oslo.messaging-tox-py35-func-kafka/633cb96/ : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-dsvm-full-rabbit/07bc6d9/ : SUCCESS in 1h 22m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/5e77be8/ : SUCCESS in 1h 35m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/fa8a936/ : SUCCESS in 1h 36m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/b50bf00/ : SUCCESS in 1h 29m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/c2cddd6/ : FAILURE in 1h 47m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-grenade-dsvm/79a2529/ : SUCCESS in 54m 51s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/90/660390/1/check/oslo.messaging-src-grenade-dsvm-multinode/f1378b4/ : SUCCESS in 1h 08m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/90/660390/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/1bab78c/ : FAILURE in 31m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/90/660390/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/308e7be/ : FAILURE in 41m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/90/660390/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/672bd13/ : FAILURE in 42m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/90/660390/1/check/openstack-tox-lower-constraints/4dd895c/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0fca44d098241320ba89cb92d05d4ff0d7e8da2f","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-22 12:49:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"676b48c916cd32e219182045a5d8d8c5c05e8837","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-06 09:09:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a2ad6221bcc1b473c76b8760d0afe2b4d281ab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:09:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcdb411a3cbd052dd58e050dc9573a5bc8f75862","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:03:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/660390/1/gate/openstack-tox-pep8/145cab3/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/90/660390/1/gate/openstack-tox-py27/6c6b5a5/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/90/660390/1/gate/openstack-tox-py35/0951929/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/90/660390/1/gate/openstack-tox-docs/648a9c1/html/ : SUCCESS in 3m 15s\n- requirements-check http://logs.openstack.org/90/660390/1/gate/requirements-check/ac98255/ : SUCCESS in 3m 14s\n- tempest-full http://logs.openstack.org/90/660390/1/gate/tempest-full/1fab3a4/ : SUCCESS in 1h 29m 23s\n- build-openstack-releasenotes http://logs.openstack.org/90/660390/1/gate/build-openstack-releasenotes/cef217d/html/ : SUCCESS in 3m 05s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/90/660390/1/gate/oslo.messaging-tox-py27-func-rabbit/70199b4/ : SUCCESS in 4m 47s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/90/660390/1/gate/oslo.messaging-src-dsvm-full-rabbit/ef311fa/ : SUCCESS in 1h 30m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/660390/1/gate/openstack-tox-lower-constraints/46570d3/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cbe40e93cc9544eaed5ca50ca754bbae0e1a7cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 11:03:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab03435ad0c108e52747c52c2cebce19b59e4c1d","revisions":{"ab03435ad0c108e52747c52c2cebce19b59e4c1d":{"kind":"REWORK","_number":1,"created":"2019-05-21 13:32:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/90/660390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/90/660390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/660390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/660390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/90/660390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/90/660390/1"}}},"commit":{"parents":[{"commit":"d75eba02c74d309cd769a8257d2d20f735c60ad1","subject":"Retry to declare a queue after internal error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d75eba02c74d309cd769a8257d2d20f735c60ad1"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-05-13 14:55:23.000000000","tz":-240},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-05-21 13:32:43.000000000","tz":120},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start\ncapping it there as well.\n\nChange-Id: Ie6b379f2c99862c37891ac03c52464e07bc2b2cc\n(cherry picked from commit 4f467fbd73092aadda8b42e517bb60a4782064ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab03435ad0c108e52747c52c2cebce19b59e4c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab03435ad0c108e52747c52c2cebce19b59e4c1d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
