)]}'
{"id":"openstack%2Foslo.messaging~667080","triplet_id":"openstack%2Foslo.messaging~stable%2Frocky~I708c3d6676b974d8daac6817c15f596cdf35817b","project":"openstack/oslo.messaging","branch":"stable/rocky","topic":"bug/1695746","hashtags":[],"change_id":"I708c3d6676b974d8daac6817c15f596cdf35817b","subject":"Issue blocking ACK for RPC requests from the consumer thread","status":"MERGED","created":"2019-06-24 13:13:19.000000000","updated":"2019-07-22 14:33:46.000000000","submitted":"2019-07-22 14:33:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667080-1563806026884-a8feb72e","meta_rev_id":"617538ba220f03865e6be8875ed850f5d7c55706","_number":667080,"virtual_id_number":667080,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-22 14:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-07-22 12:52:09.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2019-07-04 12:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2019-07-22 12:52:09.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 12:43:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-22 12:52:09.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-07-22 14:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"961ef89f2afd4c5b6113e48e6f5d93a2204e8c9a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-24 13:13:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf5bfbac4090bbb3105bb7342b828c4993dc3d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/667080/1/check/openstack-tox-pep8/ed40cd9/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/80/667080/1/check/openstack-tox-py27/26f9345/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/80/667080/1/check/openstack-tox-py35/67890d7/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/80/667080/1/check/openstack-tox-docs/0487326/html/ : SUCCESS in 3m 40s\n- tempest-full http://logs.openstack.org/80/667080/1/check/tempest-full/3a48064/ : SUCCESS in 1h 19m 26s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py27-func-rabbit/c5d219a/ : SUCCESS in 5m 19s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py27-func-amqp1/098cdcb/ : SUCCESS in 4m 36s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py27-func-kafka/4dc3e81/ : FAILURE in 4m 25s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py35-func-rabbit/276e160/ : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py35-func-amqp1/3a65250/ : SUCCESS in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-kafka http://logs.openstack.org/80/667080/1/check/oslo.messaging-tox-py35-func-kafka/41bb38b/ : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-dsvm-full-rabbit/df4c46c/ : SUCCESS in 1h 44m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/32e2f0a/ : SUCCESS in 1h 21m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/8f4e329/ : FAILURE in 1h 41m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/0bc4819/ : SUCCESS in 1h 31m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5b6d060/ : FAILURE in 1h 30m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-grenade-dsvm/f084398/ : SUCCESS in 58m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/80/667080/1/check/oslo.messaging-src-grenade-dsvm-multinode/9cd45d0/ : SUCCESS in 1h 06m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/80/667080/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/055dd09/ : FAILURE in 42m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/80/667080/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/2d3927f/ : FAILURE in 39m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/80/667080/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/9916dee/ : FAILURE in 40m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/80/667080/1/check/openstack-tox-lower-constraints/f640291/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0837a70c148416254c7a582af541d72538ebf223","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 12:43:17.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport. Can\u0027t comment too much on the change itself but this looks sensible","accounts_in_message":[],"_revision_number":1},{"id":"80b985732272b26106a3dc55a57a86208fb0a21a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-07-22 12:52:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4876d52991eaa1fa44e4deb3a7a2b6708dd8ea69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:52:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a6d08d3b342b7d8082ee0ce7be067853fc88fb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:33:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c5473dba8691a550ececa4ec282f9a02e1fe2130","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/667080/1/gate/openstack-tox-pep8/1ad80e1/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/80/667080/1/gate/openstack-tox-py27/f605d92/ : SUCCESS in 4m 59s\n- openstack-tox-py35 http://logs.openstack.org/80/667080/1/gate/openstack-tox-py35/2cc4e56/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/80/667080/1/gate/openstack-tox-docs/749b462/html/ : SUCCESS in 3m 34s\n- tempest-full http://logs.openstack.org/80/667080/1/gate/tempest-full/f85d8a7/ : SUCCESS in 1h 26m 27s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/80/667080/1/gate/oslo.messaging-tox-py27-func-rabbit/fa0bf20/ : SUCCESS in 4m 47s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/80/667080/1/gate/oslo.messaging-src-dsvm-full-rabbit/09626a9/ : SUCCESS in 1h 27m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/80/667080/1/gate/openstack-tox-lower-constraints/adb3936/ : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0648f070315b359fcf0a408eb316d9ac27adcc0","revisions":{"a0648f070315b359fcf0a408eb316d9ac27adcc0":{"kind":"REWORK","_number":1,"created":"2019-06-24 13:13:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/80/667080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/667080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/667080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/667080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/667080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/667080/1"}}},"commit":{"parents":[{"commit":"26fccea84371a3690a4d7fe970f8d6c5ba9c154b","subject":"Fix switch connection destination when a rabbitmq cluster node disappear","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/26fccea84371a3690a4d7fe970f8d6c5ba9c154b"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2018-07-10 16:03:03.000000000","tz":-240},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-06-24 13:12:34.000000000","tz":120},"subject":"Issue blocking ACK for RPC requests from the consumer thread","message":"Issue blocking ACK for RPC requests from the consumer thread\n\nThe patch for https://review.openstack.org/#/c/436958/ fixed a\nthreading problem by moving the ack back to the polling\nthread. However the RPC server expects to catch any failures of the\nACK and abort the request.  This patch adds the ACK error handling\nback to the polling thread.\n\nThis patch is based heavily off the original work done by Mehdi\nAbaakouk (sileht).\n\nChange-Id: I708c3d6676b974d8daac6817c15f596cdf35817b\nCloses-Bug: #1695746\n(cherry picked from commit 26b0be585a0b681d011edcb688750770bcdae199)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0648f070315b359fcf0a408eb316d9ac27adcc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0648f070315b359fcf0a408eb316d9ac27adcc0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
