)]}'
{"id":"openstack%2Foslo.messaging~667902","triplet_id":"openstack%2Foslo.messaging~master~Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb","project":"openstack/oslo.messaging","branch":"master","topic":"bp/transport-options","hashtags":[],"change_id":"Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb","subject":"Implement mandatory flag for RabbitMQ driver","status":"MERGED","created":"2019-06-27 11:18:12.000000000","updated":"2019-07-11 16:45:55.000000000","submitted":"2019-07-11 16:45:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667902-1562863555775-3de93a84","meta_rev_id":"117d546e387c5a32d1d46a4c918721c200709af3","_number":667902,"virtual_id_number":667902,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-11 15:16:02.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"date":"2019-07-11 15:32:52.000000000","_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-11 16:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-11 14:33:41.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":2,"date":"2019-07-03 16:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-07-11 14:33:41.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 16:48:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-11 14:33:41.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-07-11 15:16:02.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-11 16:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f1dcd20bce49f2d9ca11d8c2d3c2125eb22f464","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-27 11:18:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bd6ecc6a0242f8e6c895de3165a0e1d56046be9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:47:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/02/667902/1/check/tempest-full/63b22cb/ : FAILURE in 1h 39m 50s\n- tempest-full-py3 http://logs.openstack.org/02/667902/1/check/tempest-full-py3/a2703b2/ : SUCCESS in 1h 26m 30s\n- openstack-tox-cover http://logs.openstack.org/02/667902/1/check/openstack-tox-cover/d86b879/cover/ : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/667902/1/check/openstack-tox-lower-constraints/234e8ae/ : FAILURE in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/02/667902/1/check/openstack-tox-pep8/2487942/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/02/667902/1/check/openstack-tox-py27/1105e1e/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/02/667902/1/check/openstack-tox-py36/8a6c2a3/ : SUCCESS in 5m 08s\n- openstack-tox-py37 http://logs.openstack.org/02/667902/1/check/openstack-tox-py37/83bf723/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/02/667902/1/check/openstack-tox-docs/e9fdd1a/html/ : SUCCESS in 3m 30s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py27-func-rabbit/1ecf84a/ : SUCCESS in 5m 14s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py27-func-amqp1/cf0a75c/ : FAILURE in 11m 29s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py27-func-kafka/2dd7ae5/ : SUCCESS in 5m 34s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py36-func-rabbit/3904899/ : SUCCESS in 7m 27s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py36-func-amqp1/6af2f05/ : FAILURE in 11m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/02/667902/1/check/oslo.messaging-tox-py36-func-kafka/40e77ae/ : SUCCESS in 5m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-dsvm-full-rabbit/287aef4/ : SUCCESS in 1h 49m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/f6ab6af/ : SUCCESS in 2h 00m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/6d46a26/ : SUCCESS in 2h 21m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/9dbb2bc/ : SUCCESS in 2h 00m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/9a2e857/ : SUCCESS in 2h 10m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-grenade-dsvm/1fbfc61/ : SUCCESS in 1h 00m 07s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/02/667902/1/check/oslo.messaging-src-grenade-dsvm-multinode/3e32174/ : SUCCESS in 1h 12m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/02/667902/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/5bb4530/ : FAILURE in 47m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/02/667902/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/932b86a/ : FAILURE in 46m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/02/667902/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/6fd3f53/ : FAILURE in 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68b09aba80034b450d031be11ed571af947c6e34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-27 17:05:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51d8c667c99e6ca7f43fafeaa618cc60e3212453","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-06-27 17:29:37.000000000","message":"Patch Set 1:\n\nWeird...\nhttp://logs.openstack.org/02/667902/1/check/openstack-tox-lower-constraints/234e8ae/job-output.txt.gz#_2019-06-27_11_28_16_527807\n\nWe use kombu\u003d\u003d4.6.3 (https://github.com/celery/kombu/tree/568dcd4189d5ceeeb5bcd7bea05dcfc9a98846b5)\n\nEverything seem good in your implementation... you define the on_return on producer init [1] and when we will pass mandatory through transport_options when we will publish [2] so on_return will be called if something went wrong.\n\nI guess we need to adapt some tests [3] to avoid this `revive` [4] issue.\n\n[1] https://kombu.readthedocs.io/en/1.0/reference/kombu.messaging.html\n[2] https://kombu.readthedocs.io/en/1.0/reference/kombu.messaging.html#kombu.messaging.Producer.publish\n[3] https://github.com/openstack/oslo.messaging/blob/6cdd4cb00779acd5b5825b5f00635af08f92a5df/oslo_messaging/tests/drivers/test_impl_rabbit.py#L108\n[4] https://github.com/celery/kombu/blob/568dcd4189d5ceeeb5bcd7bea05dcfc9a98846b5/kombu/messaging.py#L232","accounts_in_message":[],"_revision_number":1},{"id":"db130d82c12f4a90b84ef1b2d638c53a82541a34","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-28 17:42:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0e8dedb3ebb8e2062087934d8595b52c29151f2","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-28 17:54:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5dabb83aacb4ad7017f98994d60df12841cfd6d0","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-28 17:57:14.000000000","message":"Patch Set 3:\n\n@Hervé you are right. \nThe error you see on the test is this [1] that I fixed here [2]\nso we need Kombu 4.6.1.\n\n1-https://github.com/celery/kombu/issues/1050\n2-https://github.com/celery/kombu/pull/1053","accounts_in_message":[],"_revision_number":3},{"id":"4405086a41a2455140ad8b7d65312d6a4ca1b5ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 19:58:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/02/667902/3/check/requirements-check/7b0a638/ : SUCCESS in 3m 45s\n- tempest-full http://logs.openstack.org/02/667902/3/check/tempest-full/2c713c9/ : SUCCESS in 1h 42m 30s\n- tempest-full-py3 http://logs.openstack.org/02/667902/3/check/tempest-full-py3/4cc49d8/ : SUCCESS in 1h 39m 04s\n- openstack-tox-cover http://logs.openstack.org/02/667902/3/check/openstack-tox-cover/034e42d/cover/ : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/667902/3/check/openstack-tox-lower-constraints/3ccf8ac/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/02/667902/3/check/openstack-tox-pep8/46918b2/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/02/667902/3/check/openstack-tox-py27/f4c3db8/ : FAILURE in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/02/667902/3/check/openstack-tox-py36/cb501e9/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/02/667902/3/check/openstack-tox-py37/5a2c27d/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/02/667902/3/check/openstack-tox-docs/851f93d/html/ : SUCCESS in 3m 04s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-rabbit/79edfe1/ : SUCCESS in 5m 06s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-amqp1/25efe07/ : FAILURE in 11m 35s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-kafka/dd1bbd7/ : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-rabbit/a02d9c6/ : SUCCESS in 5m 04s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-amqp1/e9988cf/ : FAILURE in 11m 19s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-kafka/42d2925/ : SUCCESS in 5m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-rabbit/ae219ee/ : SUCCESS in 1h 35m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/7d376d9/ : SUCCESS in 1h 46m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/e523a36/ : SUCCESS in 1h 58m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/9a424b4/ : SUCCESS in 1h 44m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/a9c3b63/ : SUCCESS in 1h 43m 12s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-grenade-dsvm/b75d5ee/ : SUCCESS in 55m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-grenade-dsvm-multinode/60dbd9c/ : SUCCESS in 1h 11m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/ed7452b/ : FAILURE in 44m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/5d12c55/ : FAILURE in 51m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/3242fd8/ : FAILURE in 47m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad9ddac4a83f0ca1e45eebf4d3cce0dcec26ee02","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-06-29 05:44:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dd84d612e549e15704b41ba956c0ef463993e490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 08:34:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/02/667902/3/check/requirements-check/b21dfc1/ : SUCCESS in 3m 28s\n- tempest-full http://logs.openstack.org/02/667902/3/check/tempest-full/625df80/ : SUCCESS in 2h 01m 07s\n- tempest-full-py3 http://logs.openstack.org/02/667902/3/check/tempest-full-py3/4263bf7/ : SUCCESS in 1h 43m 47s\n- openstack-tox-cover http://logs.openstack.org/02/667902/3/check/openstack-tox-cover/eb5b76a/cover/ : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/667902/3/check/openstack-tox-lower-constraints/99aaba5/ : SUCCESS in 6m 08s\n- openstack-tox-pep8 http://logs.openstack.org/02/667902/3/check/openstack-tox-pep8/cd06135/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/02/667902/3/check/openstack-tox-py27/c145068/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/02/667902/3/check/openstack-tox-py36/5bdfa0e/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/02/667902/3/check/openstack-tox-py37/20e0ecc/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/02/667902/3/check/openstack-tox-docs/ab8175d/html/ : SUCCESS in 4m 16s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-rabbit/7fe73a8/ : SUCCESS in 5m 43s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-amqp1/49a5177/ : FAILURE in 15m 12s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py27-func-kafka/11bfad1/ : SUCCESS in 5m 02s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-rabbit/310a81f/ : SUCCESS in 5m 24s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-amqp1/c336aa8/ : FAILURE in 10m 56s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-tox-py36-func-kafka/f63fc2a/ : FAILURE in 6m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-rabbit/239d38c/ : SUCCESS in 1h 51m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/5e11ead/ : SUCCESS in 1h 31m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/229ee14/ : SUCCESS in 2h 13m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/920f580/ : SUCCESS in 1h 49m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/1da0638/ : SUCCESS in 2h 13m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-grenade-dsvm/757b6fb/ : SUCCESS in 1h 05m 48s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/02/667902/3/check/oslo.messaging-src-grenade-dsvm-multinode/add67f1/ : SUCCESS in 1h 08m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/1008c8c/ : FAILURE in 47m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/247a7e4/ : FAILURE in 44m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/02/667902/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/f123fff/ : FAILURE in 58m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8913a495b880970c3f0176b850cd1470fb1dee94","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-07-01 15:05:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60d5743b132dca0b7e486c43a82f9264bb2d90d3","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-01 19:39:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b6964a0d48406c6a6508d71d8f7c1ed01102cc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 22:04:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/02/667902/4/check/requirements-check/a5a0540/ : SUCCESS in 30m 57s\n- tempest-full http://logs.openstack.org/02/667902/4/check/tempest-full/411fdac/ : SUCCESS in 1h 55m 34s\n- tempest-full-py3 http://logs.openstack.org/02/667902/4/check/tempest-full-py3/7f681e5/ : SUCCESS in 1h 26m 10s\n- openstack-tox-cover http://logs.openstack.org/02/667902/4/check/openstack-tox-cover/d837145/cover/ : SUCCESS in 5m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/667902/4/check/openstack-tox-lower-constraints/d93cebe/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/02/667902/4/check/openstack-tox-pep8/b5bc8c5/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/02/667902/4/check/openstack-tox-py27/58db732/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/02/667902/4/check/openstack-tox-py36/e06b396/ : SUCCESS in 9m 51s\n- openstack-tox-py37 http://logs.openstack.org/02/667902/4/check/openstack-tox-py37/243c294/ : SUCCESS in 12m 53s\n- openstack-tox-docs http://logs.openstack.org/02/667902/4/check/openstack-tox-docs/d22f303/html/ : SUCCESS in 3m 07s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py27-func-rabbit/50eebba/ : SUCCESS in 5m 06s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py27-func-amqp1/572c492/ : FAILURE in 11m 22s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py27-func-kafka/2c9080e/ : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py36-func-rabbit/16d9ab1/ : SUCCESS in 5m 43s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py36-func-amqp1/0175eb8/ : FAILURE in 12m 08s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/02/667902/4/check/oslo.messaging-tox-py36-func-kafka/8e8200f/ : SUCCESS in 11m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-dsvm-full-rabbit/d32ea7d/ : SUCCESS in 2h 10m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/d35c9ea/ : SUCCESS in 2h 16m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/5f5700b/ : SUCCESS in 2h 14m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-dsvm-full-kafka-hybrid/9045751/ : SUCCESS in 2h 21m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-dsvm-full-kafka-centos-7/67d2f17/ : FAILURE in 2h 04m 08s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-grenade-dsvm/279de73/ : SUCCESS in 1h 08m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/02/667902/4/check/oslo.messaging-src-grenade-dsvm-multinode/a1c5069/ : SUCCESS in 1h 40m 57s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/02/667902/4/check/oslo.messaging-telemetry-dsvm-integration-rabbit/b725b9e/ : FAILURE in 40m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/02/667902/4/check/oslo.messaging-telemetry-dsvm-integration-amqp1/df176e1/ : FAILURE in 40m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/02/667902/4/check/oslo.messaging-telemetry-dsvm-integration-kafka/fa6c2f5/ : FAILURE in 1h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8fd85497b5f2e1bf2422ab78640aa838cb3c7639","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-02 09:01:11.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks.\nJust a +1 for now, I need to play with your github tests before sending my +2.","accounts_in_message":[],"_revision_number":4},{"id":"305e6a424e70f449032df2df23c0298401d550fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-03 16:48:21.000000000","message":"Patch Set 4: Code-Review+2\n\nTested successfully with https://github.com/Gsantomaggio/rabbitmq-utils/tree/master/openstack/mandatory_test!\n\nReally good job!\n\nThanks for all your contribs!","accounts_in_message":[],"_revision_number":4},{"id":"a3a32d6dbf84e26d3f06758d5039716204fa8ce7","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-07-11 14:33:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nNice!","accounts_in_message":[],"_revision_number":4},{"id":"fc5964f28bed5b105d6deee342d4b241dbed7520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:33:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9aa0ce7b08c7aef4d935e0be314506c23589849f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-11 15:16:02.000000000","message":"Patch Set 4:\n\nNice, do we have docs to point people at to explain how to use this?\n\nAlso, have we considered best practices for enabling this feature? If it only works for rabbit, can consumers write their software assuming this feature? If they do, they\u0027ll essentially lose support for the other messaging backends, right?","accounts_in_message":[],"_revision_number":4},{"id":"26db7e36b705fc793750f50a81e43cf63e9c80db","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-11 15:32:52.000000000","message":"Patch Set 4:\n\n\u003e Nice, do we have docs to point people at to explain how to use this?\nThank you.\nCurrently, there is this [1] on my repository. The idea is to move it somewhere else inside OpenStack. \n\n\u003e Also, have we considered the best practices for enabling this feature? \nI will add a section inside the documentation about that. At the moment here [1]. I think that client side it should be enabled by default. \n\n\u003e If it only works for rabbit, can consumers write their software assuming this feature? \nYes, at the moment it works only in RabbitMQ.  \n\n\u003e If they do, they\u0027ll essentially lose support for the other messaging backends, right?\nThe idea is that eventually, each driver will implement this feature. \nBut no, they won\u0027t lose the support of the other backends, at the moment the feature does not work. But the feature does not impact the other backends. \n\n\n1 - https://github.com/Gsantomaggio/rabbitmq-utils/tree/master/openstack/mandatory_test","accounts_in_message":[],"_revision_number":4},{"id":"feaef353dc8e712c04921a0eae1269c170c80b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:45:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cebbd9b3c56f11d179affa649c5b98f1a19f21db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:45:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/02/667902/4/gate/requirements-check/205ef9c/ : SUCCESS in 3m 10s\n- tempest-full http://logs.openstack.org/02/667902/4/gate/tempest-full/c0633d6/ : SUCCESS in 1h 48m 04s\n- tempest-full-py3 http://logs.openstack.org/02/667902/4/gate/tempest-full-py3/94eb5e5/ : SUCCESS in 1h 50m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/02/667902/4/gate/openstack-tox-lower-constraints/458d2c5/ : SUCCESS in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/02/667902/4/gate/openstack-tox-pep8/8ebe032/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/02/667902/4/gate/openstack-tox-py27/8886b98/ : SUCCESS in 7m 02s\n- openstack-tox-py36 http://logs.openstack.org/02/667902/4/gate/openstack-tox-py36/75f5ed8/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/02/667902/4/gate/openstack-tox-py37/dab8458/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/02/667902/4/gate/openstack-tox-docs/3930ae4/html/ : SUCCESS in 3m 06s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/02/667902/4/gate/oslo.messaging-tox-py27-func-rabbit/54d66a5/ : SUCCESS in 6m 01s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/02/667902/4/gate/oslo.messaging-src-dsvm-full-rabbit/8d77c01/ : SUCCESS in 2h 05m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c50076b4efb79cef46d618d6d80eecbcebb72898","revisions":{"e903d8f0e29fbff888d866bea79e3f1b714d32b7":{"kind":"REWORK","_number":1,"created":"2019-06-27 11:18:12.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/02/667902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/02/667902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/1"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-27 11:13:47.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-27 11:18:10.000000000","tz":120},"subject":"WIP: Implement mandatory flag for RabbitMQ driver","message":"WIP: Implement mandatory flag for RabbitMQ driver\n\nWith this feature it is possible to use the mandatory RabbitMQ\nmandatory flag.\n\nImplements: blueprint transport-options ( point 3)\n\nThe blueprint link is [1]\n\nNote: this is a Work in progress, do not merge\n\nPlase follow the link [2] to use and test the feature.\n\n1- https://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e903d8f0e29fbff888d866bea79e3f1b714d32b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e903d8f0e29fbff888d866bea79e3f1b714d32b7"}]},"branch":"refs/heads/master"},"b9dad629bda667ab297d9b75984e2aee607e16f0":{"kind":"REWORK","_number":2,"created":"2019-06-28 17:42:54.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/02/667902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/02/667902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/2"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-27 11:13:47.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-28 17:41:15.000000000","tz":120},"subject":"Implement mandatory flag for RabbitMQ driver","message":"Implement mandatory flag for RabbitMQ driver\n\nWith this feature it is possible to use the mandatory RabbitMQ\nmandatory flag.\n\nImplements: blueprint transport-options ( point 3)\n\nThe blueprint link is [1]\n\nNote: this is a Work in progress, do not merge\n\nPlease follow the link [2] to use and test the feature.\n\n1- https://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b9dad629bda667ab297d9b75984e2aee607e16f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b9dad629bda667ab297d9b75984e2aee607e16f0"}]},"branch":"refs/heads/master"},"74e2895cfad14edc7f6dcf81fc0f5b9795eca0ce":{"kind":"REWORK","_number":3,"created":"2019-06-28 17:54:19.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/02/667902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/02/667902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/3"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-27 11:13:47.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-28 17:53:57.000000000","tz":120},"subject":"Implement mandatory flag for RabbitMQ driver","message":"Implement mandatory flag for RabbitMQ driver\n\nWith this feature it is possible to use the mandatory RabbitMQ\nmandatory flag.\n\nImplements: blueprint transport-options ( point 3)\n\nThe blueprint link is [1]\n\nPlease follow the link [2] to use and test the feature.\n\n1- https://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/74e2895cfad14edc7f6dcf81fc0f5b9795eca0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/74e2895cfad14edc7f6dcf81fc0f5b9795eca0ce"}]},"branch":"refs/heads/master"},"c50076b4efb79cef46d618d6d80eecbcebb72898":{"kind":"REWORK","_number":4,"created":"2019-07-01 19:39:13.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/02/667902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/02/667902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/02/667902/4"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-06-27 11:13:47.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-01 19:38:32.000000000","tz":120},"subject":"Implement mandatory flag for RabbitMQ driver","message":"Implement mandatory flag for RabbitMQ driver\n\nWith this feature it is possible to use the mandatory RabbitMQ\nmandatory flag.\n\nImplements: blueprint transport-options (point 3)\n\nThe blueprint link is [1]\n\nPlease follow the link [2] to use and test the feature.\n\n1- https://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Ie269fc08ba80c4b94a24a8207c1e86c19c3b3fcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c50076b4efb79cef46d618d6d80eecbcebb72898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c50076b4efb79cef46d618d6d80eecbcebb72898"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
