)]}'
{"id":"openstack%2Foslo.messaging~669158","triplet_id":"openstack%2Foslo.messaging~master~Ib2ba62a642e6ce45c23568daeef9703a647707f3","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1789177","hashtags":[],"change_id":"Ib2ba62a642e6ce45c23568daeef9703a647707f3","subject":"Use default exchange for direct messaging","status":"MERGED","created":"2019-07-04 12:12:06.000000000","updated":"2019-07-26 09:35:06.000000000","submitted":"2019-07-26 09:35:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669158-1564133706702-c3a5ce02","meta_rev_id":"0b1dc81354bfabae341875de28c1389049c3a7d4","_number":669158,"virtual_id_number":669158,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-26 09:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-22 13:01:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":2,"date":"2019-07-26 07:36:52.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":1,"date":"2019-07-26 07:36:52.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 12:13:01.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-04 14:40:49.000000000","updated_by":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"reviewer":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"state":"REVIEWER"},{"updated":"2019-07-22 13:01:24.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-07-26 07:36:52.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-26 09:35:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2981d9f08fe6a89bcbbf96ac7a82dcbf686db85","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-07-04 12:12:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"915f6a3036dfc8a8b17ec0253980e12bdb4f1c2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:28:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/58/669158/1/check/tempest-full/7a81714/ : SUCCESS in 1h 47m 16s\n- tempest-full-py3 http://logs.openstack.org/58/669158/1/check/tempest-full-py3/156a8ef/ : SUCCESS in 1h 37m 21s\n- openstack-tox-cover http://logs.openstack.org/58/669158/1/check/openstack-tox-cover/e5e148e/cover/ : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/669158/1/check/openstack-tox-lower-constraints/b47aed6/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/58/669158/1/check/openstack-tox-pep8/d3a384f/ : FAILURE in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/58/669158/1/check/openstack-tox-py27/961198c/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/58/669158/1/check/openstack-tox-py36/a0556c1/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/58/669158/1/check/openstack-tox-py37/786af2b/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/58/669158/1/check/openstack-tox-docs/bdce2ce/html/ : SUCCESS in 3m 08s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py27-func-rabbit/947c46b/ : SUCCESS in 5m 19s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py27-func-amqp1/8adf2b2/ : FAILURE in 17m 52s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py27-func-kafka/f7dad92/ : SUCCESS in 5m 36s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py36-func-rabbit/d9de701/ : SUCCESS in 7m 25s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py36-func-amqp1/0defabb/ : FAILURE in 13m 33s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/58/669158/1/check/oslo.messaging-tox-py36-func-kafka/314c03a/ : SUCCESS in 5m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-dsvm-full-rabbit/16b122f/ : SUCCESS in 1h 37m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e57f90b/ : SUCCESS in 1h 50m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/651f7e0/ : SUCCESS in 2h 06m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/19f3e28/ : SUCCESS in 1h 42m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/87f8794/ : SUCCESS in 2h 07m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-grenade-dsvm/0505d69/ : SUCCESS in 1h 02m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/58/669158/1/check/oslo.messaging-src-grenade-dsvm-multinode/ff7f2d4/ : SUCCESS in 1h 08m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/58/669158/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/1c38470/ : FAILURE in 42m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/58/669158/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/9629d9b/ : FAILURE in 51m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/58/669158/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/ee8eea4/ : FAILURE in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bf23d75cdb160b04f0f8b416f44743020d1c0c4","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-04 14:40:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cdaf0f058e1add15eb57e95a6c247993114cda4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-07-04 15:47:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03e988d14b1787ed5ac562bc7a4903fe908f0310","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2019-07-04 15:47:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3324e65c0a0e7a95191994b1638e86323a498365","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:06:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/58/669158/2/check/tempest-full/f27e0a9/ : SUCCESS in 1h 51m 14s\n- tempest-full-py3 http://logs.openstack.org/58/669158/2/check/tempest-full-py3/564b3c9/ : SUCCESS in 1h 42m 40s\n- openstack-tox-cover http://logs.openstack.org/58/669158/2/check/openstack-tox-cover/b806e15/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/669158/2/check/openstack-tox-lower-constraints/d0f4e96/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/58/669158/2/check/openstack-tox-pep8/affa66f/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/58/669158/2/check/openstack-tox-py27/743d378/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/58/669158/2/check/openstack-tox-py36/d594770/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/58/669158/2/check/openstack-tox-py37/b378604/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/58/669158/2/check/openstack-tox-docs/8bbe77f/html/ : SUCCESS in 3m 21s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py27-func-rabbit/e8d5c5c/ : SUCCESS in 6m 29s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py27-func-amqp1/94f68b6/ : FAILURE in 13m 18s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py27-func-kafka/8a825c8/ : SUCCESS in 7m 04s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py36-func-rabbit/da75171/ : SUCCESS in 5m 06s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py36-func-amqp1/584000b/ : FAILURE in 12m 28s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/58/669158/2/check/oslo.messaging-tox-py36-func-kafka/d3b7854/ : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-dsvm-full-rabbit/172053e/ : SUCCESS in 2h 10m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e3e687c/ : SUCCESS in 1h 47m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/7e527c8/ : SUCCESS in 1h 55m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/74e18be/ : SUCCESS in 1h 57m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/66c5e9b/ : SUCCESS in 2h 08m 43s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-grenade-dsvm/bd6b6a2/ : SUCCESS in 57m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/58/669158/2/check/oslo.messaging-src-grenade-dsvm-multinode/06c00d1/ : SUCCESS in 1h 04m 50s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/58/669158/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/9930dca/ : FAILURE in 57m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/58/669158/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/e66f2d8/ : FAILURE in 57m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/58/669158/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/a94cdcd/ : FAILURE in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b648264197f81f135b562ffbf56f01a0dc2da445","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-07-22 13:01:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b85e0f0766f54cd7b35ea50835492101109dda8b","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-26 07:36:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6194e16b3c244e950aaa58433dfd51dda282694a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 07:37:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abc1f476dae431493698e5c86fb388db3f4f845b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:35:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/58/669158/2/gate/tempest-full/4b052db/ : SUCCESS in 1h 33m 36s\n- tempest-full-py3 http://logs.openstack.org/58/669158/2/gate/tempest-full-py3/f056578/ : SUCCESS in 1h 26m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/669158/2/gate/openstack-tox-lower-constraints/00da0ba/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/58/669158/2/gate/openstack-tox-pep8/fbe8fe3/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/58/669158/2/gate/openstack-tox-py27/b1ce23f/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/58/669158/2/gate/openstack-tox-py36/e3e9bda/ : SUCCESS in 6m 12s\n- openstack-tox-py37 http://logs.openstack.org/58/669158/2/gate/openstack-tox-py37/3412c58/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/58/669158/2/gate/openstack-tox-docs/31e65ad/html/ : SUCCESS in 4m 57s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/58/669158/2/gate/oslo.messaging-tox-py27-func-rabbit/412dd0d/ : SUCCESS in 6m 24s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/58/669158/2/gate/oslo.messaging-src-dsvm-full-rabbit/066c262/ : SUCCESS in 1h 41m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2a047a1433c17ab860b23396b9aa9c55eeef5ad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:35:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6fe1aec1c74f112db297cd727d2ea400a292b038","revisions":{"fea3b1d9ada563ff1e4fd8a79762ae3ade8cafca":{"kind":"REWORK","_number":1,"created":"2019-07-04 12:12:06.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/58/669158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/669158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/1"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-07-04 12:08:45.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-07-04 12:10:44.000000000","tz":240},"subject":"Use default exchange for direct messaging","message":"Use default exchange for direct messaging\n\nLots of exchanges create problems during failover under high\nload. Please see bug report for details.\n\nThis is a step 2 patch.\n\nStep 1 was: only using default exchange\nwhen publishing.\nStep 2 is to update consumers to only listen on default exchange,\nhappening now in T release.\n\nChange-Id: Ib2ba62a642e6ce45c23568daeef9703a647707f3\nCloses-Bug: #1789177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fea3b1d9ada563ff1e4fd8a79762ae3ade8cafca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fea3b1d9ada563ff1e4fd8a79762ae3ade8cafca"}]},"branch":"refs/heads/master"},"6fe1aec1c74f112db297cd727d2ea400a292b038":{"kind":"REWORK","_number":2,"created":"2019-07-04 15:47:38.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/58/669158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/669158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/669158/2"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-07-04 12:08:45.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2019-07-04 15:47:25.000000000","tz":240},"subject":"Use default exchange for direct messaging","message":"Use default exchange for direct messaging\n\nLots of exchanges create problems during failover under high\nload. Please see bug report for details.\n\nThis is a step 2 patch.\n\nStep 1 was: only using default exchange\nwhen publishing.\nStep 2 is to update consumers to only listen on default exchange,\nhappening now in T release.\n\nChange-Id: Ib2ba62a642e6ce45c23568daeef9703a647707f3\nCloses-Bug: #1789177\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6fe1aec1c74f112db297cd727d2ea400a292b038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6fe1aec1c74f112db297cd727d2ea400a292b038"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
