)]}'
{"id":"openstack%2Foslo.messaging~669237","triplet_id":"openstack%2Foslo.messaging~master~I648dd4d4a98a3d423c98d28c193737cffe1a2c49","project":"openstack/oslo.messaging","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I648dd4d4a98a3d423c98d28c193737cffe1a2c49","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","status":"MERGED","created":"2019-07-05 02:18:15.000000000","updated":"2019-07-18 09:35:03.000000000","submitted":"2019-07-18 09:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669237-1563442503144-ef2d6f53","meta_rev_id":"2b589e80d3030cb80bb5762e2267f2abfc10a7fc","_number":669237,"virtual_id_number":669237,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-18 09:35:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-18 09:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-07-18 09:28:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 09:28:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-18 09:35:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d75ef5d48a309473eba84ee799ced5cbbdd710fc","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 02:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52418f1b6963c00bdb0a6c67604ea40c8fd1e465","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 04:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/37/669237/1/check/requirements-check/d81a2f7/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/37/669237/1/check/openstack-tox-pep8/67bae60/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/37/669237/1/check/openstack-tox-docs/be90c58/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/37/669237/1/check/build-openstack-releasenotes/133c8c6/html/ : SUCCESS in 3m 55s\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/37/669237/1/check/oslo.messaging-src-grenade-dsvm/91cf4a5/ : SUCCESS in 55m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/37/669237/1/check/oslo.messaging-src-grenade-dsvm-multinode/bd83989/ : SUCCESS in 1h 19m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/37/669237/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/f26557e/ : FAILURE in 41m 04s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/37/669237/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/750ff23/ : FAILURE in 41m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/37/669237/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/76c1237/ : FAILURE in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"583d6219a1bfef4a6a6f20ed944de7ef9330d671","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-18 09:28:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c956f9aa4ae27a7c8a338676ec5a3dec008dcd61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 09:28:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47810221d90d86865200bf9579ad37797303d527","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 09:35:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/37/669237/1/gate/requirements-check/fd73169/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/37/669237/1/gate/openstack-tox-pep8/661a634/ : SUCCESS in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/37/669237/1/gate/openstack-tox-docs/55e4de6/html/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/37/669237/1/gate/build-openstack-releasenotes/5fd0792/html/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2f40b0f3c53438df57ad68a5becf28b93fa4f17a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 09:35:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"515a2d79ef4beff37f40c52abe14678f9357ecfa","revisions":{"515a2d79ef4beff37f40c52abe14678f9357ecfa":{"kind":"REWORK","_number":1,"created":"2019-07-05 02:18:15.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/37/669237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/669237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/669237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/669237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/669237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/669237/1"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:16:45.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:16:48.000000000","tz":480},"subject":"Blacklist sphinx 2.1.0 (autodoc bug)","message":"Blacklist sphinx 2.1.0 (autodoc bug)\n\nSee https://github.com/sphinx-doc/sphinx/issues/6440 for upstream details\nDepend-On: https://review.opendev.org/#/c/663060/\n\nChange-Id: I648dd4d4a98a3d423c98d28c193737cffe1a2c49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/515a2d79ef4beff37f40c52abe14678f9357ecfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/515a2d79ef4beff37f40c52abe14678f9357ecfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
