)]}'
{"id":"openstack%2Foslo.messaging~669238","triplet_id":"openstack%2Foslo.messaging~master~Ib5e22f6a5374f05e576bbc00a209209fdb09acad","project":"openstack/oslo.messaging","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Ib5e22f6a5374f05e576bbc00a209209fdb09acad","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-07-05 02:21:10.000000000","updated":"2019-07-22 15:12:22.000000000","submitted":"2019-07-22 15:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669238-1563808342875-a91269ad","meta_rev_id":"c147060ad2f3828ddb62d690bb64932ec69f7508","_number":669238,"virtual_id_number":669238,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-22 15:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-07-22 13:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2019-07-19 13:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2019-07-22 13:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 13:40:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-22 13:23:45.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-07-22 15:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fe7cf3bf283585f96d3e1a0aa52db61329d0a12","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 02:21:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b45ada54d4109caf5f4d166d0f14cc4319333a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 04:42:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/38/669238/1/check/requirements-check/6e8f515/ : SUCCESS in 3m 28s\n- tempest-full http://logs.openstack.org/38/669238/1/check/tempest-full/d4dc2c5/ : SUCCESS in 1h 35m 07s\n- tempest-full-py3 http://logs.openstack.org/38/669238/1/check/tempest-full-py3/c2977ee/ : SUCCESS in 1h 41m 57s\n- openstack-tox-cover http://logs.openstack.org/38/669238/1/check/openstack-tox-cover/c0691d4/cover/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/669238/1/check/openstack-tox-lower-constraints/03cd23a/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/38/669238/1/check/openstack-tox-pep8/968d0a7/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/38/669238/1/check/openstack-tox-py27/0025385/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/38/669238/1/check/openstack-tox-py36/e6a9c36/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/38/669238/1/check/openstack-tox-py37/cad58a1/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/38/669238/1/check/openstack-tox-docs/0a341fe/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/38/669238/1/check/build-openstack-releasenotes/dd81c06/html/ : SUCCESS in 3m 21s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py27-func-rabbit/1326820/ : SUCCESS in 5m 45s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py27-func-amqp1/07b39c1/ : FAILURE in 11m 05s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py27-func-kafka/cb94ffc/ : SUCCESS in 5m 42s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py36-func-rabbit/c98c85e/ : SUCCESS in 5m 20s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py36-func-amqp1/0c35290/ : FAILURE in 11m 15s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/38/669238/1/check/oslo.messaging-tox-py36-func-kafka/dc9cd50/ : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-dsvm-full-rabbit/7a959c4/ : SUCCESS in 1h 22m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/f299e7b/ : SUCCESS in 1h 41m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/9b67dbb/ : SUCCESS in 2h 09m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3da15e0/ : SUCCESS in 1h 35m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/5cd5fdb/ : SUCCESS in 2h 15m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-grenade-dsvm/bdd8a86/ : SUCCESS in 57m 16s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/38/669238/1/check/oslo.messaging-src-grenade-dsvm-multinode/ae93dc8/ : SUCCESS in 1h 09m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/38/669238/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/6eac1b6/ : FAILURE in 45m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/38/669238/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/9ca0e5f/ : FAILURE in 45m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/38/669238/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/d6b9aa8/ : FAILURE in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f223f3e8e62b7a55244aacb88571c0122bee0994","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-17 10:42:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b40efb90d04ab1c76ec36b5ce79fc516fb9fba54","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-18 01:49:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"170106328ba79fa8f6a622695b34167687cff10c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 04:07:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/38/669238/2/check/requirements-check/c8bcbf2/ : SUCCESS in 3m 44s\n- tempest-full http://logs.openstack.org/38/669238/2/check/tempest-full/cffbf72/ : FAILURE in 1h 23m 39s\n- tempest-full-py3 http://logs.openstack.org/38/669238/2/check/tempest-full-py3/8a8b5e9/ : SUCCESS in 1h 26m 18s\n- openstack-tox-cover http://logs.openstack.org/38/669238/2/check/openstack-tox-cover/f67222e/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/669238/2/check/openstack-tox-lower-constraints/6773b3d/ : SUCCESS in 8m 09s\n- openstack-tox-pep8 http://logs.openstack.org/38/669238/2/check/openstack-tox-pep8/a9ec0d7/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/38/669238/2/check/openstack-tox-py27/2d36b7f/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/38/669238/2/check/openstack-tox-py36/e44832e/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/38/669238/2/check/openstack-tox-py37/6d0a62d/ : SUCCESS in 7m 53s\n- openstack-tox-docs http://logs.openstack.org/38/669238/2/check/openstack-tox-docs/73bb43c/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/38/669238/2/check/build-openstack-releasenotes/078d794/html/ : SUCCESS in 4m 04s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py27-func-rabbit/d0269ec/ : SUCCESS in 6m 57s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py27-func-amqp1/52c5adf/ : FAILURE in 12m 10s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py27-func-kafka/2f936f2/ : SUCCESS in 5m 39s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py36-func-rabbit/58c35b1/ : SUCCESS in 6m 11s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py36-func-amqp1/dc8efe9/ : FAILURE in 12m 12s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/38/669238/2/check/oslo.messaging-tox-py36-func-kafka/8894c85/ : SUCCESS in 7m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-dsvm-full-rabbit/94f1068/ : SUCCESS in 1h 38m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/11078e3/ : SUCCESS in 1h 46m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/78ecd0a/ : FAILURE in 1h 38m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/b7d8c11/ : SUCCESS in 1h 43m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/ff43076/ : SUCCESS in 1h 50m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-grenade-dsvm/8984520/ : SUCCESS in 1h 08m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/38/669238/2/check/oslo.messaging-src-grenade-dsvm-multinode/7008ba6/ : SUCCESS in 1h 11m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/38/669238/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/f935886/ : FAILURE in 48m 33s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/38/669238/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/2b915f1/ : FAILURE in 46m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/38/669238/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/3022255/ : FAILURE in 48m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1357425e59f93c68f78f2dbde058bbbbc984ab7d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-18 08:46:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d673fa46e3a4aa0acb5f4cadc907de220e5f98db","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-19 03:24:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4568ccac2f2da014e130565297e91816ca66c38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 05:34:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/38/669238/3/check/requirements-check/52ea061/ : SUCCESS in 3m 01s\n- tempest-full http://logs.openstack.org/38/669238/3/check/tempest-full/48ccbec/ : SUCCESS in 1h 41m 59s\n- tempest-full-py3 http://logs.openstack.org/38/669238/3/check/tempest-full-py3/2088f42/ : SUCCESS in 1h 17m 37s\n- openstack-tox-cover http://logs.openstack.org/38/669238/3/check/openstack-tox-cover/50b2cb3/cover/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/669238/3/check/openstack-tox-lower-constraints/9fab5d2/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/38/669238/3/check/openstack-tox-pep8/6526702/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/38/669238/3/check/openstack-tox-py27/91c40e7/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/38/669238/3/check/openstack-tox-py36/4044355/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/38/669238/3/check/openstack-tox-py37/1c5b965/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/38/669238/3/check/openstack-tox-docs/2efbf25/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/38/669238/3/check/build-openstack-releasenotes/9625aee/html/ : SUCCESS in 3m 12s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py27-func-rabbit/6e03644/ : SUCCESS in 4m 24s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py27-func-amqp1/32c0e84/ : FAILURE in 11m 03s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py27-func-kafka/d2eaf03/ : SUCCESS in 4m 56s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py36-func-rabbit/70d16e0/ : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py36-func-amqp1/df82adc/ : FAILURE in 10m 53s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/38/669238/3/check/oslo.messaging-tox-py36-func-kafka/4e604f0/ : SUCCESS in 5m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-dsvm-full-rabbit/4212ec9/ : SUCCESS in 1h 28m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/e29d7a5/ : SUCCESS in 1h 32m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/60bf9af/ : SUCCESS in 1h 56m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-dsvm-full-kafka-hybrid/adf428c/ : SUCCESS in 1h 30m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-dsvm-full-kafka-centos-7/70a637a/ : SUCCESS in 2h 03m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-grenade-dsvm/870aeb0/ : SUCCESS in 1h 04m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/38/669238/3/check/oslo.messaging-src-grenade-dsvm-multinode/71a1a5f/ : SUCCESS in 1h 08m 32s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/38/669238/3/check/oslo.messaging-telemetry-dsvm-integration-rabbit/0aad53c/ : FAILURE in 40m 55s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/38/669238/3/check/oslo.messaging-telemetry-dsvm-integration-amqp1/871abd5/ : FAILURE in 39m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/38/669238/3/check/oslo.messaging-telemetry-dsvm-integration-kafka/0186489/ : FAILURE in 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9abf0312a80a39297fcff78c2d639f53e3039d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-19 13:40:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"51fc79a7a34455127722ed6fc5926c6312135028","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-07-22 13:23:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dfc83681f7a6bdca8f7858694b19c216a551ed69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:23:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f5456ccf0fb0b760fe51b3367e46f06988df1c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:12:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/38/669238/3/gate/requirements-check/ad27d23/ : SUCCESS in 3m 45s\n- tempest-full http://logs.openstack.org/38/669238/3/gate/tempest-full/22fb7e9/ : SUCCESS in 1h 34m 04s\n- tempest-full-py3 http://logs.openstack.org/38/669238/3/gate/tempest-full-py3/6bc6e90/ : SUCCESS in 1h 23m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/669238/3/gate/openstack-tox-lower-constraints/6cddab2/ : SUCCESS in 6m 16s\n- openstack-tox-pep8 http://logs.openstack.org/38/669238/3/gate/openstack-tox-pep8/d64d188/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/38/669238/3/gate/openstack-tox-py27/a8411aa/ : SUCCESS in 6m 20s\n- openstack-tox-py36 http://logs.openstack.org/38/669238/3/gate/openstack-tox-py36/c6c398e/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/38/669238/3/gate/openstack-tox-py37/dbfe045/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/38/669238/3/gate/openstack-tox-docs/c21fc28/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/38/669238/3/gate/build-openstack-releasenotes/05882b0/html/ : SUCCESS in 3m 57s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/38/669238/3/gate/oslo.messaging-tox-py27-func-rabbit/4a04da4/ : SUCCESS in 6m 44s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/38/669238/3/gate/oslo.messaging-src-dsvm-full-rabbit/14f291f/ : SUCCESS in 1h 42m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4f238f990496dbbca972765ba9c3accd0e47662d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:12:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33dc529c378c64864cb22f75f31f620cfd151be0","revisions":{"afcac345b17f150b8e7298cf5b06a2612e52ea3c":{"kind":"REWORK","_number":1,"created":"2019-07-05 02:21:10.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/38/669238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/669238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/1"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:20:23.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:20:27.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ib5e22f6a5374f05e576bbc00a209209fdb09acad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afcac345b17f150b8e7298cf5b06a2612e52ea3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afcac345b17f150b8e7298cf5b06a2612e52ea3c"}]},"branch":"refs/heads/master"},"f239114fde7d80b6695e1465a55d27f7dcda9019":{"kind":"REWORK","_number":2,"created":"2019-07-18 01:49:06.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/38/669238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/669238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/2"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:20:23.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-18 01:49:02.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ib5e22f6a5374f05e576bbc00a209209fdb09acad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f239114fde7d80b6695e1465a55d27f7dcda9019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f239114fde7d80b6695e1465a55d27f7dcda9019"}]},"branch":"refs/heads/master"},"33dc529c378c64864cb22f75f31f620cfd151be0":{"kind":"REWORK","_number":3,"created":"2019-07-19 03:24:39.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/38/669238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/669238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/669238/3"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 02:20:23.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-19 03:24:08.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ib5e22f6a5374f05e576bbc00a209209fdb09acad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/33dc529c378c64864cb22f75f31f620cfd151be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/33dc529c378c64864cb22f75f31f620cfd151be0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
