)]}'
{"id":"openstack%2Foslo.messaging~669462","triplet_id":"openstack%2Foslo.messaging~master~I699a3ee210c4b45018566b520eaf641e0e582362","project":"openstack/oslo.messaging","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I699a3ee210c4b45018566b520eaf641e0e582362","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:47:54.000000000","updated":"2019-07-11 13:23:41.000000000","submitted":"2019-07-11 13:23:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669462-1562851421786-77280bbb","meta_rev_id":"3a01c3a58c8bb64da9a0b776406ea17a08fcd8ee","_number":669462,"virtual_id_number":669462,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-11 13:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-07-08 14:30:14.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-11 11:22:04.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-07-11 11:22:04.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 14:30:14.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-11 11:22:04.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-11 13:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04529ba4252e29fcfa0ab6212870875ed5d12f6f","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:47:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce61d9b93e9962844a85f54ecc79d1de6d6d3360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/669462/1/check/test-release-openstack/23d51f9/ : SUCCESS in 2m 01s\n- tempest-full http://logs.openstack.org/62/669462/1/check/tempest-full/6aab216/ : SUCCESS in 1h 45m 05s\n- tempest-full-py3 http://logs.openstack.org/62/669462/1/check/tempest-full-py3/4bde385/ : SUCCESS in 1h 30m 52s\n- openstack-tox-cover http://logs.openstack.org/62/669462/1/check/openstack-tox-cover/4995b16/cover/ : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/669462/1/check/openstack-tox-lower-constraints/288bc26/ : SUCCESS in 8m 17s\n- openstack-tox-pep8 http://logs.openstack.org/62/669462/1/check/openstack-tox-pep8/089ce74/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/62/669462/1/check/openstack-tox-py27/9c8164c/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/62/669462/1/check/openstack-tox-py36/a4e8205/ : SUCCESS in 7m 26s\n- openstack-tox-py37 http://logs.openstack.org/62/669462/1/check/openstack-tox-py37/33b7ded/ : SUCCESS in 8m 11s\n- openstack-tox-docs http://logs.openstack.org/62/669462/1/check/openstack-tox-docs/feb454d/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/62/669462/1/check/build-openstack-releasenotes/dcd70c9/html/ : SUCCESS in 4m 59s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py27-func-rabbit/16c9177/ : SUCCESS in 5m 11s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py27-func-amqp1/dd868c6/ : FAILURE in 11m 24s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py27-func-kafka/5b89301/ : SUCCESS in 4m 59s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py36-func-rabbit/a1822ba/ : SUCCESS in 5m 17s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py36-func-amqp1/98573aa/ : FAILURE in 11m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/62/669462/1/check/oslo.messaging-tox-py36-func-kafka/fcee12a/ : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-dsvm-full-rabbit/d5d377c/ : SUCCESS in 1h 37m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/44695de/ : SUCCESS in 1h 41m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/0330d67/ : SUCCESS in 2h 01m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/4e27f5b/ : SUCCESS in 1h 44m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/0004114/ : SUCCESS in 2h 06m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-grenade-dsvm/5e44ac1/ : SUCCESS in 59m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/62/669462/1/check/oslo.messaging-src-grenade-dsvm-multinode/36fc1c6/ : SUCCESS in 1h 10m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/62/669462/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/ce6568e/ : FAILURE in 43m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/62/669462/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/5a5437f/ : FAILURE in 40m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/62/669462/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/e83a668/ : FAILURE in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a18f036417d238c48f193971794187a6e6feecaf","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-08 14:30:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5f5f2aa8f399285d9bb393c1eb4cb11b8b2fb38","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-11 11:22:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26eaa6b81595ded56daf6d442e4a4592bdda3b12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:22:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41f9665973023a626d80bf2d4627852b0c4502e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:23:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a06892a2d7be958bc214b525fcb8af0882d76a74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:23:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/669462/1/gate/test-release-openstack/95a3496/ : SUCCESS in 2m 07s\n- tempest-full http://logs.openstack.org/62/669462/1/gate/tempest-full/393bc4f/ : SUCCESS in 1h 46m 26s\n- tempest-full-py3 http://logs.openstack.org/62/669462/1/gate/tempest-full-py3/2d35d4e/ : SUCCESS in 1h 49m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/669462/1/gate/openstack-tox-lower-constraints/3f75e25/ : SUCCESS in 7m 43s\n- openstack-tox-pep8 http://logs.openstack.org/62/669462/1/gate/openstack-tox-pep8/c3f0086/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/62/669462/1/gate/openstack-tox-py27/b2fa860/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/62/669462/1/gate/openstack-tox-py36/9bc662f/ : SUCCESS in 4m 43s\n- openstack-tox-py37 http://logs.openstack.org/62/669462/1/gate/openstack-tox-py37/e0ac294/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/62/669462/1/gate/openstack-tox-docs/5b08718/html/ : SUCCESS in 3m 13s\n- build-openstack-releasenotes http://logs.openstack.org/62/669462/1/gate/build-openstack-releasenotes/8b243e0/html/ : SUCCESS in 3m 47s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/62/669462/1/gate/oslo.messaging-tox-py27-func-rabbit/00e7d6d/ : SUCCESS in 7m 18s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/62/669462/1/gate/oslo.messaging-src-dsvm-full-rabbit/65a95b0/ : SUCCESS in 1h 55m 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c31abc7bc7b1b7dacd8c482857ab8829ed05eaf","revisions":{"1c31abc7bc7b1b7dacd8c482857ab8829ed05eaf":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:47:54.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/62/669462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/62/669462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/669462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/669462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/62/669462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/62/669462/1"}}},"commit":{"parents":[{"commit":"6cdd4cb00779acd5b5825b5f00635af08f92a5df","subject":"Merge \"Implement the transport options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6cdd4cb00779acd5b5825b5f00635af08f92a5df"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:51.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:47:53.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I699a3ee210c4b45018566b520eaf641e0e582362\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c31abc7bc7b1b7dacd8c482857ab8829ed05eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c31abc7bc7b1b7dacd8c482857ab8829ed05eaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
