)]}'
{"id":"openstack%2Foslo.messaging~672073","triplet_id":"openstack%2Foslo.messaging~master~Iac7474c06ef425a2afe5bcd912e51510ba1c8fb3","project":"openstack/oslo.messaging","branch":"master","topic":"bp/transport-options","hashtags":[],"change_id":"Iac7474c06ef425a2afe5bcd912e51510ba1c8fb3","subject":"Add the mandatory flag for direct send","status":"MERGED","created":"2019-07-22 15:34:22.000000000","updated":"2019-08-26 15:07:02.000000000","submitted":"2019-08-26 15:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672073-1566831912229-b3eddfb8","meta_rev_id":"6355ade59654d9e05680a0a30fd219a2fc1fb42e","_number":672073,"virtual_id_number":672073,"owner":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2019-08-26 15:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-08-22 17:47:46.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-08-26 13:14:53.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-08-26 13:14:53.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 13:19:42.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-23 13:20:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-24 09:53:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-08-22 17:47:46.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-08-26 13:14:53.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-08-26 15:05:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"784c944af8f227dd5a0a967fbd752c8bb59e8a2b","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-22 15:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c36ec49b02ceaffc7b8a3d5544f7e68a660381d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/672073/1/check/tempest-full/ef503c3/ : SUCCESS in 1h 28m 16s\n- tempest-full-py3 http://logs.openstack.org/73/672073/1/check/tempest-full-py3/24b6908/ : SUCCESS in 1h 22m 35s\n- openstack-tox-cover http://logs.openstack.org/73/672073/1/check/openstack-tox-cover/d4e9a5a/cover/ : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/672073/1/check/openstack-tox-lower-constraints/4872014/ : SUCCESS in 6m 17s\n- openstack-tox-pep8 http://logs.openstack.org/73/672073/1/check/openstack-tox-pep8/523ffe9/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/73/672073/1/check/openstack-tox-py27/72e25a3/ : SUCCESS in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/73/672073/1/check/openstack-tox-py36/19db8b6/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/73/672073/1/check/openstack-tox-py37/7ba86d5/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/73/672073/1/check/openstack-tox-docs/3d9a6e7/html/ : SUCCESS in 3m 12s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py27-func-rabbit/753afa5/ : SUCCESS in 7m 08s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py27-func-amqp1/79fb604/ : FAILURE in 12m 46s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py27-func-kafka/a6576d3/ : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py36-func-rabbit/31c70a4/ : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py36-func-amqp1/68903db/ : FAILURE in 12m 12s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/672073/1/check/oslo.messaging-tox-py36-func-kafka/05450d6/ : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-dsvm-full-rabbit/76929e2/ : SUCCESS in 1h 54m 40s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/34ba6c5/ : SUCCESS in 1h 33m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/d151777/ : FAILURE in 1h 47m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/3d05a72/ : SUCCESS in 1h 47m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/68cf7f0/ : SUCCESS in 1h 49m 24s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-grenade-dsvm/b158f75/ : SUCCESS in 1h 02m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/672073/1/check/oslo.messaging-src-grenade-dsvm-multinode/bbb7010/ : SUCCESS in 1h 21m 27s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/672073/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/dbdb172/ : FAILURE in 51m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/672073/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/5912ca6/ : FAILURE in 50m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/672073/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/6e83718/ : FAILURE in 55m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86263a89fe8a3a3220c61278b1cda43876fb05fe","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-23 13:01:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71e08b94e4bf859cb842be2e7d522c666c4a59bb","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-07-23 13:18:04.000000000","message":"Patch Set 2:\n\nI\u0027d like to add the tests for this functionality, but I don\u0027t know how to do that. \nTo raise the MessageUndeliverable exception, the `reply_q` [1] \nshould not exist.\nThe `reply_q` is automatically created by the library, and can\u0027t find a way to disable it.\n\nany idea?\n\n\n1- https://github.com/openstack/oslo.messaging/blob/master/oslo_messaging/_drivers/amqpdriver.py#L123","accounts_in_message":[],"_revision_number":2},{"id":"7db029f125187c38d22f57913f6463d057308c48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:33:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/73/672073/2/check/tempest-full/28bf421/ : SUCCESS in 1h 45m 19s\n- tempest-full-py3 http://logs.openstack.org/73/672073/2/check/tempest-full-py3/2482518/ : SUCCESS in 1h 41m 28s\n- openstack-tox-cover http://logs.openstack.org/73/672073/2/check/openstack-tox-cover/c4ec40e/cover/ : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/672073/2/check/openstack-tox-lower-constraints/0a86fbb/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/73/672073/2/check/openstack-tox-pep8/f545347/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/73/672073/2/check/openstack-tox-py27/15e756c/ : SUCCESS in 6m 56s\n- openstack-tox-py36 http://logs.openstack.org/73/672073/2/check/openstack-tox-py36/2922056/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/73/672073/2/check/openstack-tox-py37/caab0bd/ : SUCCESS in 10m 04s\n- openstack-tox-docs http://logs.openstack.org/73/672073/2/check/openstack-tox-docs/1d17b0c/html/ : SUCCESS in 4m 06s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py27-func-rabbit/6947d0c/ : SUCCESS in 6m 18s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py27-func-amqp1/c1e0df8/ : FAILURE in 11m 23s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py27-func-kafka/0ad42a5/ : FAILURE in 6m 58s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py36-func-rabbit/8232a63/ : SUCCESS in 6m 38s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py36-func-amqp1/77e7b8d/ : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/73/672073/2/check/oslo.messaging-tox-py36-func-kafka/0449e5b/ : SUCCESS in 6m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-dsvm-full-rabbit/ede2233/ : SUCCESS in 2h 30m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/3a9ba45/ : SUCCESS in 1h 33m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/b44b15d/ : SUCCESS in 2h 10m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-dsvm-full-kafka-hybrid/163f684/ : SUCCESS in 1h 34m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-dsvm-full-kafka-centos-7/ba3f57b/ : SUCCESS in 2h 09m 43s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-grenade-dsvm/0e2aa75/ : SUCCESS in 57m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/73/672073/2/check/oslo.messaging-src-grenade-dsvm-multinode/7aabd5c/ : SUCCESS in 1h 11m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/73/672073/2/check/oslo.messaging-telemetry-dsvm-integration-rabbit/55fe982/ : FAILURE in 43m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/73/672073/2/check/oslo.messaging-telemetry-dsvm-integration-amqp1/06ac3e4/ : FAILURE in 47m 41s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/73/672073/2/check/oslo.messaging-telemetry-dsvm-integration-kafka/7520cc2/ : FAILURE in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8033aba04758b8a516d39b673069ea7bb267ef9b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-24 09:47:40.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"f00a4305f6893a6d91d6a82569c87894125ee854","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-24 09:53:27.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e I\u0027d like to add the tests for this functionality, but I don\u0027t know\n \u003e how to do that.\n \u003e To raise the MessageUndeliverable exception, the `reply_q` [1]\n \u003e should not exist.\n \u003e The `reply_q` is automatically created by the library, and can\u0027t\n \u003e find a way to disable it.\n \u003e \n \u003e any idea?\n \u003e \n \u003e \n \u003e 1- https://github.com/openstack/oslo.messaging/blob/master/oslo_messaging/_drivers/amqpdriver.py#L123\n\nMaybe you need to using the functional tests instead unit tests let the server and driver do things normally and in parallel in the test removing the routing_key and after resend a message by using the driver.\n\nI move from +2 to 0 during the reflection phase related to the previous lines.","accounts_in_message":[],"_revision_number":2},{"id":"2414df3e0c89fefe7de95003df9bdf1fab067048","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-08-05 13:55:08.000000000","message":"Patch Set 2:\n\nHmmm - yeah that\u0027s a good point.\n\nTotally untesting/only partially thought through suggestion:  what if you create a functional test that defines an RPC method that causes the server to sleep(N) seconds and have the client\u0027s RPC call time out after 1 second?   Will that cause the queue to be torn down before the reply is generated?","accounts_in_message":[],"_revision_number":2},{"id":"153416be01d24a808a7e4b6d4dd15228b3e10a07","author":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"date":"2019-08-21 12:47:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9eefa4fb6ac1569582a822d28e8cbf40973b4fbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:39:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/209139d678154f148e971cbfc543b351 : SUCCESS in 1h 34m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68625402cb184799bf1fad32e6cf3799 : SUCCESS in 1h 22m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/accf06c60ded42049f1141ce6c76d822 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/827c2d9012b4490dacb226e598f40023 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852e9fb205624fc48857e05be96da970 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9e6a68726354f67beac9a679564ed67 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d643a034424d8bbb4da63dee900512 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89663ca39d7d4fe7bbbf949a874421c4 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/459abdbe99ca414aadbd6df1752e689c : SUCCESS in 3m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/681a8db8621e4d7a893ab4e81c89a40e : SUCCESS in 8m 54s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/985c2c4260ae4f9499fec0880b70d7bd : FAILURE in 13m 41s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/9c5985758f074f5ea7f8fc4e3ae936cf : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/8ad7888083844132b57933dabce9cc8e : SUCCESS in 7m 11s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/56c456a36981436fa073a78ce7e10f05 : FAILURE in 13m 47s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/8b3ede24ff0244aeadbe9e98c293ad43 : FAILURE in 6m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/75df6d97b81540c39ddf7de25786c6c5 : SUCCESS in 1h 39m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/eea7ab9b45854333bf586d6a119d8b50 : SUCCESS in 1h 42m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/a0dff30ff39644e082b6ad80565a4604 : SUCCESS in 2h 24m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5661a257a896411f99d77aba1c8d94b5 : FAILURE in 1h 34m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/450390c8871f4c568fd34ea40fdbcd05 : SUCCESS in 2h 10m 35s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/4e39dd53287a421e987b4da45676eace : SUCCESS in 59m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/113c9c982aa744da9c94eb57fc283b72 : FAILURE in 1h 15m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/daa73183841646e59f68733eff44e12a : FAILURE in 59m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/bcdfacc08c144674a823deab8f019185 : FAILURE in 46m 59s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d765bed992ec4e8a8a713d95f762e79f : FAILURE in 54m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3846bb341304cfb89442f24a1aa71b2b6d68ce7c","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-08-22 17:47:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bfffec3f66a42e6c13d19d5740530ae8ea66578a","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-26 13:14:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"167e28fca6e81ddf1421ea85407baa8362c8e6a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 13:15:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"74aa88b5cd415ce71590e17a828e631fa38ed922","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:05:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/33fff4e3babf4850a77b646ff6bec251 : SUCCESS in 1h 34m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec2a359c2d0f4ced864f160e43f927ac : SUCCESS in 1h 25m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a47990cbdc8d45f6bfe30d8ac6905d3e : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c63b7825fc1340e39e6eb5acea7f2a67 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/787ed982c3f74a21ad7501083806f3f0 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705393598add4d7c803a7953040f1592 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d379873c0a8345da818aa4c9b9972f01 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe5b8871fb148dca0ce652c7ee9fc86 : SUCCESS in 5m 51s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/545df1e3af824450a5fd1843bdbb25be : SUCCESS in 8m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/711c106a7c764c89a9fcf97d8cad3436 : SUCCESS in 1h 43m 44s","accounts_in_message":[],"_revision_number":3},{"id":"7ba9677da891b94ccd36796a4364926d3d672491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1490b1058aeee78c251f9f0a82b16090a0f1c48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:07:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3229343bedee4af280a05dc90158e1da : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b7e9faf6590086b79f9696183b5c296ecc12b7b6","revisions":{"546fa2b65383f6e69486bdcab6f699489cf9c6e6":{"kind":"REWORK","_number":1,"created":"2019-07-22 15:34:22.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/672073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/672073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/1"}}},"commit":{"parents":[{"commit":"cbdc5df3eb97fdfbe97602b80c49beaf263aa604","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cbdc5df3eb97fdfbe97602b80c49beaf263aa604"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-22 15:27:21.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-22 15:34:19.000000000","tz":120},"subject":"WIP Add the mandatory flag for direct send","message":"WIP Add the mandatory flag for direct send\n\nWith this feature, the server will raise and log a Message Undeliverable\nexception. So it is possible to log immediately an error in case the\nreply queue does not exist for some reason.\n\nThis is part of blueprint transport-options\nThe blueprint link is [1]\n\nPlease follow the link [2] to use and test the feature.\n\nNote: This is a WIP, so please don\u0027t merge.\n\n1-\nhttps://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Iac7474c06ef425a2afe5bcd912e51510ba1c8fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/546fa2b65383f6e69486bdcab6f699489cf9c6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/546fa2b65383f6e69486bdcab6f699489cf9c6e6"}]},"branch":"refs/heads/master"},"7e8bd5c63ea54a7f415a66b3d821b1540a9fb946":{"kind":"REWORK","_number":2,"created":"2019-07-23 13:01:11.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/672073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/672073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/2"}}},"commit":{"parents":[{"commit":"cbdc5df3eb97fdfbe97602b80c49beaf263aa604","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cbdc5df3eb97fdfbe97602b80c49beaf263aa604"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-22 15:27:21.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-23 13:00:39.000000000","tz":120},"subject":"Add the mandatory flag for direct send","message":"Add the mandatory flag for direct send\n\nWith this feature, the server will raise and log a Message Undeliverable\nexception. So it is possible to log immediately an error in case the\nreply queue does not exist for some reason.\n\nThis is part of blueprint transport-options\nThe blueprint link is [1]\n\nPlease follow the link [2] to use and test the feature.\n\n1-\nhttps://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Iac7474c06ef425a2afe5bcd912e51510ba1c8fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e8bd5c63ea54a7f415a66b3d821b1540a9fb946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e8bd5c63ea54a7f415a66b3d821b1540a9fb946"}]},"branch":"refs/heads/master"},"b7e9faf6590086b79f9696183b5c296ecc12b7b6":{"kind":"REWORK","_number":3,"created":"2019-08-21 12:47:24.000000000","uploader":{"_account_id":29364,"name":"Gabriele Santomaggio","email":"g.santomaggio@gmail.com","username":"gsantomaggio"},"ref":"refs/changes/73/672073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/73/672073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/73/672073/3"}}},"commit":{"parents":[{"commit":"b56380654a4b695a8a27827621b185ee3991ce92","subject":"Introduce RabbitMQ driver documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b56380654a4b695a8a27827621b185ee3991ce92"}]}],"author":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-07-22 15:27:21.000000000","tz":120},"committer":{"name":"Gabriele","email":"gsantomaggio@suse.com","date":"2019-08-21 12:46:59.000000000","tz":120},"subject":"Add the mandatory flag for direct send","message":"Add the mandatory flag for direct send\n\nWith this feature, the server will raise and log a Message Undeliverable\nexception. So it is possible to log immediately an error in case the\nreply queue does not exist for some reason.\n\nThis is part of blueprint transport-options\nThe blueprint link is [1]\nPlease follow the link [2] to use and test the feature.\n\n1-\nhttps://blueprints.launchpad.net/oslo.messaging/+spec/transport-options\n2- https://github.com/Gsantomaggio/rabbitmq-utils/\ntree/master/openstack/mandatory_test\n\nChange-Id: Iac7474c06ef425a2afe5bcd912e51510ba1c8fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b7e9faf6590086b79f9696183b5c296ecc12b7b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b7e9faf6590086b79f9696183b5c296ecc12b7b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
