)]}'
{"id":"openstack%2Foslo.messaging~673514","triplet_id":"openstack%2Foslo.messaging~master~I4a80d8c6b162a99176eadb052f6201dc38dbc5f9","project":"openstack/oslo.messaging","branch":"master","topic":"fix-acknowledge-spell","hashtags":[],"change_id":"I4a80d8c6b162a99176eadb052f6201dc38dbc5f9","subject":"Correcting typo in acknowledge spelling","status":"MERGED","created":"2019-07-30 13:16:14.000000000","updated":"2019-07-30 17:37:34.000000000","submitted":"2019-07-30 17:37:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673514-1564508254649-bedbf1ba","meta_rev_id":"1c61d1f9d2f8568877f2f5573fbf87537eadda07","_number":673514,"virtual_id_number":673514,"owner":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-30 17:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-30 14:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-07-30 14:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 14:05:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-30 17:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"410865fc12aeb5ce976917fb94bd81e3bed27e69","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2019-07-30 13:16:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27e5b054fb030b75a8bbb45f9926ff63d0fb8e81","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-30 14:05:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45a8e6e07369bc3e352120ced092e001e537bfc3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-30 14:05:28.000000000","message":"Patch Set 1:\n\nYou might need to fix your git configuration. I assume even your nickname isn\u0027t \u0027root\u0027 :)","accounts_in_message":[],"_revision_number":1},{"id":"1d606f32d6c59fb7455793cc7d9e22661c689a80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/14/673514/1/check/tempest-full/8397d00/ : SUCCESS in 1h 36m 47s\n- tempest-full-py3 http://logs.openstack.org/14/673514/1/check/tempest-full-py3/a8ceae2/ : SUCCESS in 1h 32m 53s\n- openstack-tox-cover http://logs.openstack.org/14/673514/1/check/openstack-tox-cover/3912417/cover/ : SUCCESS in 6m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/673514/1/check/openstack-tox-lower-constraints/2c33df2/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/14/673514/1/check/openstack-tox-pep8/654b94f/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/14/673514/1/check/openstack-tox-py27/b12ef16/ : SUCCESS in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/14/673514/1/check/openstack-tox-py36/885c331/ : SUCCESS in 6m 02s\n- openstack-tox-py37 http://logs.openstack.org/14/673514/1/check/openstack-tox-py37/bcc3a52/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/14/673514/1/check/openstack-tox-docs/7babd20/html/ : SUCCESS in 4m 51s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py27-func-rabbit/c252758/ : SUCCESS in 6m 34s\n- oslo.messaging-tox-py27-func-amqp1 http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py27-func-amqp1/38c3f1a/ : FAILURE in 11m 33s (non-voting)\n- oslo.messaging-tox-py27-func-kafka http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py27-func-kafka/b0d60cc/ : SUCCESS in 6m 28s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py36-func-rabbit/0dfcc57/ : FAILURE in 7m 59s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py36-func-amqp1/c4b9b29/ : FAILURE in 11m 29s (non-voting)\n- oslo.messaging-tox-py36-func-kafka http://logs.openstack.org/14/673514/1/check/oslo.messaging-tox-py36-func-kafka/4ff87d7/ : SUCCESS in 6m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-dsvm-full-rabbit/e43b416/ : SUCCESS in 1h 37m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/89ce732/ : SUCCESS in 1h 37m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/45fd53c/ : SUCCESS in 1h 56m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/f047226/ : SUCCESS in 1h 41m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/17b59a9/ : SUCCESS in 2h 00m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-grenade-dsvm/0313a20/ : SUCCESS in 1h 05m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode http://logs.openstack.org/14/673514/1/check/oslo.messaging-src-grenade-dsvm-multinode/9952716/ : SUCCESS in 1h 14m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit http://logs.openstack.org/14/673514/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/a56e9ff/ : FAILURE in 47m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 http://logs.openstack.org/14/673514/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/09bcff9/ : FAILURE in 55m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka http://logs.openstack.org/14/673514/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/577f008/ : FAILURE in 51m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"375a52b7e73b2fc86780e896eb743f182dc36df4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:24:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fefb0bfb2d17f7f915cc7b3962383a59e3e57bd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:37:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ad8add03448e8e9439da69c75c327b6920e3a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 17:37:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/14/673514/1/gate/tempest-full/6ebe343/ : SUCCESS in 2h 05m 48s\n- tempest-full-py3 http://logs.openstack.org/14/673514/1/gate/tempest-full-py3/10e8b16/ : SUCCESS in 1h 32m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/673514/1/gate/openstack-tox-lower-constraints/4029e53/ : SUCCESS in 6m 46s\n- openstack-tox-pep8 http://logs.openstack.org/14/673514/1/gate/openstack-tox-pep8/cd0bde6/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/14/673514/1/gate/openstack-tox-py27/a21edd6/ : SUCCESS in 7m 53s\n- openstack-tox-py36 http://logs.openstack.org/14/673514/1/gate/openstack-tox-py36/6498529/ : SUCCESS in 6m 11s\n- openstack-tox-py37 http://logs.openstack.org/14/673514/1/gate/openstack-tox-py37/1968ab4/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/14/673514/1/gate/openstack-tox-docs/3dc8774/html/ : SUCCESS in 5m 00s\n- oslo.messaging-tox-py27-func-rabbit http://logs.openstack.org/14/673514/1/gate/oslo.messaging-tox-py27-func-rabbit/44d371b/ : SUCCESS in 5m 49s\n- oslo.messaging-src-dsvm-full-rabbit http://logs.openstack.org/14/673514/1/gate/oslo.messaging-src-dsvm-full-rabbit/28fbe43/ : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"042caf8a53fda895eeef987a6ef22d67a7556e5e","revisions":{"042caf8a53fda895eeef987a6ef22d67a7556e5e":{"kind":"REWORK","_number":1,"created":"2019-07-30 13:16:14.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/14/673514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/14/673514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/673514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/673514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/673514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/14/673514/1"}}},"commit":{"parents":[{"commit":"002ac1453b2302b4885c4d0e37cc420bc30b4bb3","subject":"Merge \"Use default exchange for direct messaging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/002ac1453b2302b4885c4d0e37cc420bc30b4bb3"}]}],"author":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-30 07:44:59.000000000","tz":330},"committer":{"name":"root","email":"sandyada@redhat.com","date":"2019-07-30 07:47:07.000000000","tz":330},"subject":"Correcting typo in acknowledge spelling","message":"Correcting typo in acknowledge spelling\n\nThere is a typographical errors in amqpdriver.py. Correcting spelling\nfrom acknowlege to acknowledge.\n\nChange-Id: I4a80d8c6b162a99176eadb052f6201dc38dbc5f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/042caf8a53fda895eeef987a6ef22d67a7556e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/042caf8a53fda895eeef987a6ef22d67a7556e5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
