)]}'
{"id":"openstack%2Foslo.messaging~674914","triplet_id":"openstack%2Foslo.messaging~master~I4116e8871436097dea650f56e7b187358367d92e","project":"openstack/oslo.messaging","branch":"master","topic":"choice-nit","hashtags":[],"change_id":"I4116e8871436097dea650f56e7b187358367d92e","subject":"Fix nits on kafka compression help text","status":"MERGED","created":"2019-08-06 19:02:14.000000000","updated":"2019-08-07 14:43:16.000000000","submitted":"2019-08-07 14:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674914-1565188996740-62203689","meta_rev_id":"3ffb13a6e806a76370dfa119c89f7dc4f3c48eba","_number":674914,"virtual_id_number":674914,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2019-08-07 14:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2019-08-06 23:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-08-07 12:40:45.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-08-07 12:40:45.000000000","_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 23:28:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-08-07 12:40:45.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-08-07 14:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b77b08793afcc47f9488f7d9eeb65b8a27d280e6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-06 19:02:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a54e72014919c6328ec97ea0eac3807203f99d50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:27:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/14/674914/1/check/tempest-full/9ea222f/ : SUCCESS in 2h 03m 13s\n- tempest-full-py3 https://logs.opendev.org/14/674914/1/check/tempest-full-py3/63b9171/ : SUCCESS in 1h 24m 07s\n- openstack-tox-cover https://logs.opendev.org/14/674914/1/check/openstack-tox-cover/1806036/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://logs.opendev.org/14/674914/1/check/openstack-tox-lower-constraints/88a7411/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://logs.opendev.org/14/674914/1/check/openstack-tox-pep8/0a33940/ : SUCCESS in 4m 03s\n- openstack-tox-py27 https://logs.opendev.org/14/674914/1/check/openstack-tox-py27/8ea4376/ : SUCCESS in 5m 33s\n- openstack-tox-py36 https://logs.opendev.org/14/674914/1/check/openstack-tox-py36/8e060d2/ : SUCCESS in 7m 10s\n- openstack-tox-py37 https://logs.opendev.org/14/674914/1/check/openstack-tox-py37/aef7ef9/ : SUCCESS in 6m 33s\n- openstack-tox-docs https://logs.opendev.org/14/674914/1/check/openstack-tox-docs/9919408/html/ : SUCCESS in 4m 13s\n- oslo.messaging-tox-py27-func-rabbit https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py27-func-rabbit/878f04e/ : SUCCESS in 7m 15s\n- oslo.messaging-tox-py27-func-amqp1 https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py27-func-amqp1/4b47805/ : FAILURE in 12m 21s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py27-func-kafka/4e0841b/ : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py36-func-rabbit/26e1f9e/ : FAILURE in 7m 24s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py36-func-amqp1/be69193/ : FAILURE in 13m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://logs.opendev.org/14/674914/1/check/oslo.messaging-tox-py36-func-kafka/88d61dd/ : FAILURE in 7m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-dsvm-full-rabbit/22b3c49/ : SUCCESS in 1h 35m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/f0ecf61/ : SUCCESS in 1h 48m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/48b5a95/ : SUCCESS in 2h 09m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/ea8b392/ : SUCCESS in 1h 36m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/eb8668c/ : SUCCESS in 2h 16m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-grenade-dsvm/67cd466/ : FAILURE in 1h 15m 31s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://logs.opendev.org/14/674914/1/check/oslo.messaging-src-grenade-dsvm-multinode/efd7df9/ : SUCCESS in 1h 16m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://logs.opendev.org/14/674914/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/72bf113/ : FAILURE in 1h 05m 20s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://logs.opendev.org/14/674914/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/6fb95f1/ : FAILURE in 49m 19s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://logs.opendev.org/14/674914/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/7dfab6d/ : FAILURE in 45m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fe149fd416e66e7b50b8bac52226fafca0323f7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-06 23:28:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d52259c06bba57a3f05817867202fc806b54ca7","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-07 12:40:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"833f17ee9230e97b9822b2b5ab6b970ac4137128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:40:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b33f8df1b90ff4bac620eace8686c4760dc2c1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:43:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ed84b8879b7dc7260b2dbefcdaf2786849b18e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:43:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://logs.opendev.org/14/674914/1/gate/tempest-full/0df0950/ : SUCCESS in 1h 39m 18s\n- tempest-full-py3 https://logs.opendev.org/14/674914/1/gate/tempest-full-py3/abfa768/ : SUCCESS in 1h 13m 04s\n- openstack-tox-lower-constraints https://logs.opendev.org/14/674914/1/gate/openstack-tox-lower-constraints/5adee8d/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://logs.opendev.org/14/674914/1/gate/openstack-tox-pep8/8d300f6/ : SUCCESS in 3m 49s\n- openstack-tox-py27 https://logs.opendev.org/14/674914/1/gate/openstack-tox-py27/4d0182e/ : SUCCESS in 7m 36s\n- openstack-tox-py36 https://logs.opendev.org/14/674914/1/gate/openstack-tox-py36/8fc2ac8/ : SUCCESS in 5m 58s\n- openstack-tox-py37 https://logs.opendev.org/14/674914/1/gate/openstack-tox-py37/6728458/ : SUCCESS in 6m 13s\n- openstack-tox-docs https://logs.opendev.org/14/674914/1/gate/openstack-tox-docs/f61def4/html/ : SUCCESS in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit https://logs.opendev.org/14/674914/1/gate/oslo.messaging-tox-py27-func-rabbit/074df6d/ : SUCCESS in 5m 58s\n- oslo.messaging-src-dsvm-full-rabbit https://logs.opendev.org/14/674914/1/gate/oslo.messaging-src-dsvm-full-rabbit/6f77ef6/ : SUCCESS in 1h 55m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39196a4e90f0d46f6b8122b2e421b519b9c6d93f","revisions":{"39196a4e90f0d46f6b8122b2e421b519b9c6d93f":{"kind":"REWORK","_number":1,"created":"2019-08-06 19:02:14.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/14/674914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/14/674914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/674914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/674914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/674914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/14/674914/1"}}},"commit":{"parents":[{"commit":"dfc8fe9a8ff81ee45505a40ec161df9eb7f37c25","subject":"Merge \"Support kafka message compression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfc8fe9a8ff81ee45505a40ec161df9eb7f37c25"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-08-06 19:00:24.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-08-06 19:00:24.000000000","tz":0},"subject":"Fix nits on kafka compression help text","message":"Fix nits on kafka compression help text\n\nThere is no need to explicitly list the choices in the help text.\nThe oslo.config sample generator will include the choices automatically[0]\n\nAlso tweaks the wording of text to make it clear that it is the allowed\nvalues which vary based on kafka version.\n\nChange-Id: I4116e8871436097dea650f56e7b187358367d92e\n0: https://github.com/openstack/oslo.config/blob/2488c1e1cee151a3a65c60f371f0bea2f15d4bca/oslo_config/generator.py#L263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/39196a4e90f0d46f6b8122b2e421b519b9c6d93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/39196a4e90f0d46f6b8122b2e421b519b9c6d93f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
