)]}'
{"id":"openstack%2Foslo.messaging~675697","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~Id11db4113c9b1c3add602192c1e915218704ef27","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"stein","hashtags":[],"change_id":"Id11db4113c9b1c3add602192c1e915218704ef27","subject":"fix typos","status":"MERGED","created":"2019-08-09 20:26:14.000000000","updated":"2019-08-12 22:30:30.000000000","submitted":"2019-08-12 22:30:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675697-1565649030609-f1476ba1","meta_rev_id":"db3160c461619bde52a3aec717e241a3eb35250a","_number":675697,"virtual_id_number":675697,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-08-09 21:33:07.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":2,"date":"2019-08-12 22:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-08-12 20:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-08-12 12:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-08-12 20:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 12:52:57.000000000","updated_by":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"reviewer":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"state":"REVIEWER"},{"updated":"2019-08-12 20:33:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-12 22:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e26b65f753eecb70df3bcb5878a29b07c3557b56","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-09 20:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a12d3760bc1b99cd6fe4f36108d3e0fde472f77","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-09 21:33:07.000000000","message":"Patch Set 1:\n\nto avoid grep issue during debug session on stable branches...","accounts_in_message":[],"_revision_number":1},{"id":"7c080c6af2b28eda22797c5ef65be228861728b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 23:21:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://logs.opendev.org/97/675697/1/check/tempest-full/1efd8f5/ : SUCCESS in 1h 40m 41s\n- tempest-full-py3 https://logs.opendev.org/97/675697/1/check/tempest-full-py3/0fa4716/ : SUCCESS in 1h 20m 19s\n- openstack-tox-cover https://logs.opendev.org/97/675697/1/check/openstack-tox-cover/3088033/cover/ : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/97/675697/1/check/openstack-tox-lower-constraints/49cf2c9/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://logs.opendev.org/97/675697/1/check/openstack-tox-pep8/9b22549/ : SUCCESS in 3m 12s\n- openstack-tox-py27 https://logs.opendev.org/97/675697/1/check/openstack-tox-py27/f99d4d8/ : SUCCESS in 4m 27s\n- openstack-tox-py36 https://logs.opendev.org/97/675697/1/check/openstack-tox-py36/f22a25a/ : SUCCESS in 7m 00s\n- openstack-tox-py37 https://logs.opendev.org/97/675697/1/check/openstack-tox-py37/faf332b/ : SUCCESS in 6m 37s\n- openstack-tox-docs https://logs.opendev.org/97/675697/1/check/openstack-tox-docs/487796e/html/ : SUCCESS in 2m 49s\n- oslo.messaging-tox-py27-func-rabbit https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py27-func-rabbit/33bb7f2/ : SUCCESS in 6m 16s\n- oslo.messaging-tox-py27-func-amqp1 https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py27-func-amqp1/5bd1678/ : SUCCESS in 4m 01s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py27-func-kafka/6a6e235/ : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py35-func-rabbit/df424be/ : SUCCESS in 5m 52s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py35-func-amqp1/8962ef4/ : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://logs.opendev.org/97/675697/1/check/oslo.messaging-tox-py35-func-kafka/e2ef44d/ : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-dsvm-full-rabbit/8ce8d1f/ : SUCCESS in 1h 51m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-dsvm-full-amqp1-hybrid/90627cc/ : SUCCESS in 1h 56m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-dsvm-full-amqp1-centos-7/9186fe4/ : FAILURE in 1h 38m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-dsvm-full-kafka-hybrid/2be3f90/ : FAILURE in 17m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-dsvm-full-kafka-centos-7/a34848f/ : FAILURE in 29m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-grenade-dsvm/01ea112/ : SUCCESS in 49m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://logs.opendev.org/97/675697/1/check/oslo.messaging-src-grenade-dsvm-multinode/5b06108/ : SUCCESS in 1h 10m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://logs.opendev.org/97/675697/1/check/oslo.messaging-telemetry-dsvm-integration-rabbit/34c65fd/ : FAILURE in 41m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://logs.opendev.org/97/675697/1/check/oslo.messaging-telemetry-dsvm-integration-amqp1/791ad9a/ : FAILURE in 38m 54s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://logs.opendev.org/97/675697/1/check/oslo.messaging-telemetry-dsvm-integration-kafka/3353fd8/ : FAILURE in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b8a3d96628e4861ab0ca577c74cbb694ad65527","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-12 12:52:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a22965cce3afcf843e80dcac690d0a3755de57e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-12 20:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8e89f55e083c4f31816ac7d5c6392682ccb83d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 20:33:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c986dcc6190f86b62522a22d78176d3deed94ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:30:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/06659d8c85e045da82b47e4f974d2a5a : SUCCESS in 1h 50m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2825b6967a4340b29a003f3d83ed9694 : SUCCESS in 1h 38m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa765906a5944884ac6ca2c31c560250 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fa746f900349d1adaa2973cf9223cd : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56626c9701674473944b63634416fd0e : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380bce9ea41c4a7484194780405452ec : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4161681e682d470ab7b1932b062e42dc : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2d8c4c9aa24844bf7e352eb4f6f179 : SUCCESS in 2m 56s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/f6d205f11b0341b584b662b08ff93358 : SUCCESS in 4m 55s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2052a34ebfc0440a8a527fe0a6cd26a3 : SUCCESS in 1h 45m 29s","accounts_in_message":[],"_revision_number":1},{"id":"da1e51296e32ac0b5f19cf0899b5e7c779d06922","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:30:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76fce54b9e693203f268774f876fa478ab44412c","revisions":{"76fce54b9e693203f268774f876fa478ab44412c":{"kind":"REWORK","_number":1,"created":"2019-08-09 20:26:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/97/675697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/97/675697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/97/675697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/97/675697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/97/675697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/97/675697/1"}}},"commit":{"parents":[{"commit":"89c4d7ecd0a2a0cd939610d1b35e086b590f13b3","subject":"Merge \"Fix switch connection destination when a rabbitmq cluster node disappear\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89c4d7ecd0a2a0cd939610d1b35e086b590f13b3"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-06-17 15:58:47.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-08-09 19:56:40.000000000","tz":120},"subject":"fix typos","message":"fix typos\n\nChange-Id: Id11db4113c9b1c3add602192c1e915218704ef27\n(cherry picked from commit 54356899da5d4a0997abb6819f184a27929cc0e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76fce54b9e693203f268774f876fa478ab44412c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76fce54b9e693203f268774f876fa478ab44412c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
