)]}'
{"id":"openstack%2Foslo.messaging~677744","triplet_id":"openstack%2Foslo.messaging~master~I0f86416623a0b718516147f0660b4df2b74cf867","project":"openstack/oslo.messaging","branch":"master","topic":"scenario-tests","hashtags":[],"change_id":"I0f86416623a0b718516147f0660b4df2b74cf867","subject":"Setup backend scenarios for functional tests","status":"MERGED","created":"2019-08-21 14:41:03.000000000","updated":"2020-02-21 15:58:39.000000000","submitted":"2020-02-21 15:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677744-1582300346797-a5ae4dff","meta_rev_id":"94f974f89bffd1f5f67c9b34e32672e13c119df5","_number":677744,"virtual_id_number":677744,"owner":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-21 15:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2020-02-13 17:43:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-02-21 14:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-02-21 14:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 17:43:42.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2020-02-21 14:12:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-21 15:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"379945786ea22623df488fd33c0a9e808f9ae705","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-21 14:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a3514327abf096e9408f179361b721a2cd9d68e","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-21 14:41:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"858c18a0e3335f5b0d15d9cc749efeaeabff1a3b","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-21 16:06:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03498e90c5a113e7954bbe1e5e9d56560823eb6c","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-21 17:08:07.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"0144bf3b8801a0f28dd6b542051b432c38fb788b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 18:23:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8e78211dcf0a4e7d9a8e29b55c570e18 : FAILURE in 1h 27m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4efe11b17dd42d38d9a95f8490ea75a : SUCCESS in 1h 22m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e93be95e2a0d4e32b4c801abc559fe7a : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48bf0473cdfb447189138f72b37bd185 : POST_FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659a748bbec24714a86d3cad171fb39c : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86677e1ee5b54825ac7317f3196d434c : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84dde4ece37b4d9485b23a0c5fc19f22 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f1f074e403749aeb0ca46f6f497a041 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d465b086d1403bbfcf26ec8a669cd8 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1937a09e09b94dc898c8f45858533e38 : SUCCESS in 5m 20s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/2d298fa911204adc9e8278ac7b4c5231 : SUCCESS in 5m 57s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/f4dd1f55c6404cc28074dabaabf07847 : TIMED_OUT in 42m 31s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/4516c911881c44e0b299af355d426c16 : RETRY_LIMIT in 2m 47s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/f4f5236018104dd390b8c3954c21aeb9 : RETRY_LIMIT in 5m 03s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/aa06873519f94197a7a5ba138eef47e2 : SUCCESS in 8m 48s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/62b6ef9ed0f747d99f2d9311d736e9cf : TIMED_OUT in 41m 35s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/801ec0fc24a842bdb9f723868e772162 : RETRY_LIMIT in 2m 13s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/4ffa0d52d9544802bc47e058707739a7 : RETRY_LIMIT in 2m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e27410f6e01f4eb880092720552b6f8b : SUCCESS in 1h 33m 17s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6d4471ff4d344edc98b67fb6aeac578d : SUCCESS in 1h 55m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/c1a305e7ad9b455f91fa4b999b3ab8d1 : SUCCESS in 2h 13m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fed81d0f98064bf382b314cb59fb5216 : SUCCESS in 1h 46m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/694a1d3a4a6548fb85d00b110725af7a : SUCCESS in 2h 05m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/574d2bdea9c34b2ea00762f11fb065af : SUCCESS in 1h 03m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1a121cfa7e0f48ea9685a66db1d032a5 : SUCCESS in 1h 34m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/fa553bc919b344a380b7f6abed6618e9 : FAILURE in 50m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/b86f200ce87648e8adb02755fe42d032 : FAILURE in 47m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/761ebf54509c4fc5be55df7e8e2b9851 : FAILURE in 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e4be220006b6da8cddb14f747c2a71784889bda","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-26 14:12:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3fb1cb61973cf0d9da2fc777a211ba802f16b688","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-26 14:12:42.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e7cec00cd79c05346daac57045c488a35d4b294b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:25:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0dd56a911f9645aa89ce7d47acf676fc : SUCCESS in 1h 46m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2b6ecb5ec4741c4b7107df1bd1051b3 : SUCCESS in 1h 27m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c79ddf62898240d3b7b73acb5c5a9278 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fbda092fe5c40998a72f825a54c0df6 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708d005585bc436da823dc5a193154c3 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17c365fc31fe4ccb9666240698d04fb8 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ba252e30487489fbe55a3da834aa6fe : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ace8a8a3bb540b7b4958dbca83eea98 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a99e78899f489da365b9410d8aa82a : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286c7fb7d2074fda91cab1f8d00226c0 : SUCCESS in 5m 11s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/c8f64a21ad3447b7815420b7cb91cd93 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/2d410161c142410686fa1e92ef98ca07 : TIMED_OUT in 41m 51s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/5712f52baa414b7e8d7c31517d5f102d : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/051fd1170c974313be0e8f6e85ba515e : FAILURE in 6m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6fd8db91aaee4f76aaf43aabe01b6d3f : SUCCESS in 8m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/2b853a0ccc334fc8b2add76b3fb42cc3 : TIMED_OUT in 41m 32s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/5dbe1fd781c5443998c061c44ee535c1 : FAILURE in 6m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/9f548c3d93e94823a7ee2d3bbff4c836 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/446f7a3577774bfab971a6357588c435 : SUCCESS in 1h 51m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8d71ee33aa28433e9f026a66bcda2aee : SUCCESS in 1h 40m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7564784f1b584fbea42e06835267e06f : SUCCESS in 2h 03m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9f4761f10fb64df8b91f526b074aceea : SUCCESS in 1h 42m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/23dd8985aa534cc7996f2048fe6e3645 : SUCCESS in 1h 48m 53s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/fe21e95e16cd4a78a43bffc8794cd26d : SUCCESS in 1h 08m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/02066f0f90dd403a87676c5a90f23edb : SUCCESS in 1h 20m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/716d233467034a0399c78c9715f70c60 : FAILURE in 49m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d5b5537d57b04692a764a9551560c7fc : FAILURE in 47m 53s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d40d4dc7a75d4b56a361c87eb77ff2e4 : FAILURE in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a650efc4462fcbc27e833002b06a3f146e95fd3","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-08-27 15:35:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"58e3581e6d8243bd24fee7734f7a6acb1ce74283","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 17:55:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/351858f951b34283b065367647fbfaca : SUCCESS in 1h 43m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d4c2844418b4918a0161c6331c90eff : SUCCESS in 1h 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f7b62625cba4e8689a7d22805d552f7 : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7601a7d6fd694e7597bd2b40a290dd81 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e80f89e86684e83a1dd044a7e3958c7 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acb1d6ca23c04f4aaf0e6d6e47834a87 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0c6d8342c4645bdba28c2d694eca3e7 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d882cbe14aef4f079761e149e1e61033 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee149c4eec274164ac03f7d260511276 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea5255f74fb848a08a91e8dd45158412 : SUCCESS in 4m 05s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/075176fdf45b4b92b791ff6e53459173 : SUCCESS in 6m 17s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/2f529604355847239b7885773d998d41 : SUCCESS in 6m 28s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/ae0abfd5957c4cae8872d8fdb1e13f53 : FAILURE in 5m 27s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/263b8d7348214959873e94c0e9147de5 : FAILURE in 5m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/85f4e9cce402459fab048b701f5b79c8 : SUCCESS in 6m 04s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b8702a3a6fb8429e987cda231f8e01d5 : SUCCESS in 6m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/a02d5b4e076f4731aec1ab0ad887f773 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c13e07adb8cd4a5396ed17ba8f165a44 : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/525b92f7352a485b9dcb9e1acd8c1ac9 : SUCCESS in 1h 45m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7b03af38b9264dd9a208145b2b5bea3d : SUCCESS in 1h 33m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/b2bce3122776434ebd727e97e12b3d82 : SUCCESS in 2h 12m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c95cea8f3c4b4006b114553aabbd47bb : SUCCESS in 1h 41m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/8e3729ad220e435abb2b1db3f5e15f9d : SUCCESS in 2h 01m 38s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d67ff975a23d466e9130bd1e6dd26357 : SUCCESS in 1h 03m 30s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ad0656460edd46cd90362ee38404e8d9 : SUCCESS in 1h 13m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b9beede84e1c4188b543c10b4527bf80 : FAILURE in 48m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/68b09038c1cd49d7a664aff0f71c788d : FAILURE in 1h 01m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/5743348318d645c3881c5f0e56d9edd5 : FAILURE in 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b7456dcf1ab24eafa995cbbfe6116fa6cc7c8e66","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-10-07 12:46:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8bce1a5b13e9dfe34f22ae55e625898b1b29e6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 15:08:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba103a84a0254b8caede758175a40465 : SUCCESS in 1h 41m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/254342a972a9418196cb6cd67c6566aa : SUCCESS in 1h 35m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7246fe14b704a6eaa02cd20043d5ead : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80e9814643434d0d9e7a90f657b7cec6 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f205c5802e714fbc89eff542dd2daa90 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf2501439be94b38a56b526b02fa35ca : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ecbb781865e48219edf8aceb5c11a55 : TIMED_OUT in 40m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba9637cbfdf441ddaadfd3ddecc602e5 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12880d2144e4febaf2fb85218ba5a4c : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3da58a8283a84481ab980bb6c2de894b : SUCCESS in 4m 53s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/d9aecac5380342478164254684379f9f : SUCCESS in 6m 53s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/b152ce0240c041d1bbef451b0eba0eaf : SUCCESS in 6m 25s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/4f978d73af534c01a9621b4dfc7a38e2 : FAILURE in 7m 43s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/a6b84327db6241ccac2016dccdc441da : FAILURE in 6m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/dcb9af9c161a43958ad995e8a139b7c5 : SUCCESS in 5m 24s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/6db812d576874c82abf63b66ef2937e5 : SUCCESS in 6m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/42890bddf9f44dca8f3cc8ec738a7c4a : FAILURE in 5m 52s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/cc7aeb9ff27141aaab696080a98e49c9 : FAILURE in 6m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7b254fdebd324904a96ab19f60e97f12 : SUCCESS in 1h 50m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/02601eb7d6a24797b4004f794a4ad681 : SUCCESS in 1h 46m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/659c89acf9884bad85f160e2db04a440 : SUCCESS in 2h 16m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f644bb851b4e4684862e5f2265d72b04 : SUCCESS in 1h 42m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/73fe328bcdd7469f8d19f48ebbf9d709 : FAILURE in 1h 47m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/e86c3f45f69f49c9b4cf025ec51be20a : SUCCESS in 1h 05m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0187a6eb21b748058fd7d9dab51db9cf : SUCCESS in 1h 16m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/804e45781da34816ad6e01f97d684c9e : FAILURE in 50m 14s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/1339efddde374beabbfaf9c6aaded53b : FAILURE in 36m 58s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d8e97994010c4252906c78f3e14a5bf4 : FAILURE in 45m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a84336ba71f6e387866ba41796e779388745ede","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-11-04 13:52:56.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"0372c9f1daf6ad31d724f2c3230866307f55ccae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:04:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/35e548f0a7e1477bb29f43f14d9b6b3d : SUCCESS in 1h 43m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b635ed177e0d4fab97be0eaebb575ef4 : SUCCESS in 1h 12m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff6a3f2d281c4f38b5c7e77c38362c44 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38aa0960bb544e10b610a2428e9f494b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51868adb22b04fb3b3efd74ea47bbe1b : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8d81fa5e57a4c2eb84f3895cae0533d : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a67bed02bb894ca9b449eb5109c74125 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d1af5e62ac874887b91be905ca46b2ae : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93082c19b12b4cc295525406683d60ab : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e53db097550c4e1f838a1320f83bd8d1 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/1acb8a31f9ed425f9048888e8f24a5a4 : SUCCESS in 5m 31s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/91cb09b08d26482c9724a47589fbaa4b : SUCCESS in 5m 27s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/13fac1c7b749476189c197abc45d1d35 : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/6ba24a5ba85b4d15a12edcc7799b617e : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2a0b01439fe24f029e43380ef65288a0 : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/092b44c8158a43d1bf76e6d2442c5be2 : SUCCESS in 5m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/66ecbf6c6f5248bc81fb0813a4787e72 : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/b79fce84bec5431caa2d6a666608510a : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d6e87b2b5278431da22063b85cc4042e : SUCCESS in 1h 47m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2652aea1088a482b90c78c7a2e0f6297 : SUCCESS in 1h 40m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7b950e8febf647b787df6f9bdfa9a9b6 : POST_FAILURE in 1h 59m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0732a5d6c1f6496b9dfd29e7b888024e : SUCCESS in 1h 50m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b9bdc2a5a8a243ad892946aaeb2f56d0 : SUCCESS in 2h 07m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/63af4c800330412085239b65b1e294bb : SUCCESS in 1h 04m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/3f3947f74d3048718ba048610e75d52c : SUCCESS in 1h 12m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/bd280b3398974846b6c894badb71f884 : FAILURE in 48m 26s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/d8c563ece98c436ca26c92b9bc39be51 : FAILURE in 49m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/f5bf2c2bf8054bdea8ffe7fbcaad7552 : FAILURE in 1h 02m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"416c63c864f1c08f76f176c2fc3d87d7ccf090f8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-04 17:21:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nHello Andy,\n\nNice changes!\n\nI put a -1 related to the commit message which, I think, need to bring more infos.\n\nI\u0027ve some doubts, questions, and suggestions, see my comments inline:","accounts_in_message":[],"_revision_number":6},{"id":"727a65ce772cb19ebfe6348c5a96b0a1fe385c89","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-16 17:16:22.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"639eaab44809d29298e16eb67161b8c7254f7cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:19:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/90dd7498567f490e9c435f1eeaa49764 : SUCCESS in 1h 38m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af12bfca0fad40e68bcc581d668b4dc8 : SUCCESS in 1h 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c54fcc48dbc4294870fea9834580034 : SUCCESS in 7m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad01632115f443c9bbaedb276e4301c : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f787d2cf63a748c5962fa426e1c4bf42 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f86c3127a8940ac8f02088597634980 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebacf82c39724f26abc539331bcc2962 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99ca832f6ca44179896849be1a5912ee : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04930fb840914bf4936cb20175d5e40e : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee59c4c80554450aaae183b8a7a2353a : SUCCESS in 5m 56s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/5cf212d94cfd4964baab39c9a22bea4c : SUCCESS in 6m 33s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/3749cce3ef3d4ae59e53a739076ff961 : FAILURE in 9m 35s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/587079f34e6f48a5b1186f726606caa4 : FAILURE in 6m 04s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/3aa1194cf1504582b0cd16126b8c8b6c : FAILURE in 7m 50s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7f3ea278ded64cb39e4723759d744fe2 : SUCCESS in 8m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/b92aae57c0cc458cb7dc323b5bf9a0ff : SUCCESS in 8m 01s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/721dbdfdafc24f88a404d668c94d9aaf : FAILURE in 8m 31s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/29c82b8316714f75b0f1ff8bfcc1e397 : FAILURE in 6m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e8916c154a714a7b9da0578bf9ac7476 : SUCCESS in 2h 01m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/86db5677958643eab6b1ccbe58b9eeda : SUCCESS in 1h 37m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/1d72f6f979c54c6c856cae4cd61c1e88 : FAILURE in 13m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/43273a512d474344a8e99e88153b8a8a : SUCCESS in 1h 51m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/663b83afd2004e4bafbeaeb8181b6dcf : FAILURE in 10m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f02bed7385504fbfbc585e052ad440e9 : SUCCESS in 1h 11m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/283527cc80264a70bafb1f6bae6768ae : FAILURE in 1h 13m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/3a88e73254f64ed7b9106a6b5601f302 : FAILURE in 41m 10s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/f2a8087ab51a4a39bf4ef91af136fe12 : FAILURE in 52m 07s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/9424ab39b1c04a49b7f155a88a0119c4 : FAILURE in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"060e9bd345b96ed404a35f2705083e80d6434e2d","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-16 20:44:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d3d43aa42934e86ec55de6cd9b68f9664e79c232","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-16 20:44:44.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"870ef0dd62a79d4dd6e6db1d406a9ffef55c13d0","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-16 21:07:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9c4203b8b7bb55684f8e9c639f2b0e2067e87251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:13:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4de408cc28c94a7eaa458aab1c25adee : SUCCESS in 1h 24m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/447298f2c1ab48f69776ad84dc00be5b : SUCCESS in 2h 05m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f94cb4421050439d9726ad85b0197b74 : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80eb289a548145e98ec005c3e99e0a49 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421d0d5c2bb4458f8876e37511806be2 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2348517bedc348669b31ad48a035bf29 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76fe5009584b4d87839931c71f0d2615 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4bd948f3505496894a25be121d04682 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea2c1a9eb3649ea940c02867b5ea8c8 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63fde4aaade400f8a656fd4877b1e22 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/6239e10181184727a462535fc312c08d : FAILURE in 6m 27s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/8183552d019249768f499e852af591b6 : FAILURE in 7m 11s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/13dbd8116a2743fd88c0d49a1054bff3 : FAILURE in 6m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/3f14494f3cf547ec8348ea7045a3275f : FAILURE in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"791cd4976c30ff9bcde55a93ea55d242abc5dba9","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-19 17:42:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7dbad59397d829be84997d58e18d231d27d86ce1","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-19 17:43:03.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"5e811f476035d91b77d5023f69dd640aeb296596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 19:21:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a9c969246554d89b540d6596d8a9d6b : SUCCESS in 1h 37m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7df4d93ac93942179924fefbef2a9f28 : SUCCESS in 1h 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed324788dc6435984ac3ead305160b5 : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3464d55808f4205afa91b78d9eecf14 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6e9f6b91a244c5b45671a81d601501 : FAILURE in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83d7fa91ff524d11bf1f87b9b1fdbd7f : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03edc8d0eb5544cbadb7a1389cf60335 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b779152d0a6e4cb78b9f4d526c4408d6 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6240ac6cd4a440b592e15b28e41e6504 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3edb75d5528748c298af7a6eba40740e : SUCCESS in 7m 25s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/9f88186ecba949ab974edd4231c44042 : FAILURE in 7m 53s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e3ddc1ba26834712b2f30ea9b25c77b5 : FAILURE in 9m 09s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/0d94bf5c7f644919b649a1cb95e25534 : FAILURE in 7m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/d64e2c386040472b8bdacb13f741a664 : FAILURE in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2b1c0966f0950bf3b48f1b2d4300b833479fb88e","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-19 21:32:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7ddc1d1d6a64a261979f3b676c0a25e0316061b3","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-19 21:32:51.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"5a7ed12edc4678760df15c1c3fa231ecd9004517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 23:16:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d153ee1082ba48b5a0221329fc362124 : SUCCESS in 1h 43m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0cf27bd512249a59992b036bc2d7096 : SUCCESS in 1h 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35fd91ee51004a68b3a2039949040e2f : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4d04e114cc4b10af27b3bc983d9d4e : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df8abc16a9f474283d58bb24f0ab55a : FAILURE in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a41d6be28014b819f9003e7aa74aeee : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0efa786d91ed43c090fe68e20e20c79d : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6159767eb92c424ea5b4567594ced12c : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24cb5e0421e846c2aa5e6e84115bad5f : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eef8660a3eaf49eda94679b5d1fc18e4 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/3096516d116d40a68ad9ecf119b74faf : FAILURE in 6m 03s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/506588a654194e33abf86e59fe5b174f : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/ad71df1df14a4d3c9ce383bd8a37d3ab : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c9d79f2f18084603beb6ea4e42c7891a : FAILURE in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7e06edd78fab17c52d3557f2767b033c5e693566","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-20 13:31:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ccb792b8e0d1b2bdaf5783ee1b306242abacff5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 15:23:56.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd4877389eeb4403ae1b803c45ad0145 : SUCCESS in 1h 46m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86cdcffe7ed44938935d60dcf51512cb : SUCCESS in 1h 24m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d43c7ba707f7409f90a4ea9c63078506 : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/659a7894429547269c0f4e6fda750b4b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbabe8e2b3147a58fe1c3c5617a6173 : FAILURE in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dc5f2841a6f46358fa8c76f4298833a : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad965a13b5f0431087a86238ec5bc766 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cf73fc81a684bfe80652351a1bca001 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136428aa18dd419081da6c46f8e2da7d : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cf36bf7ace54593b6931064b8a37b0c : SUCCESS in 4m 22s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/10f734db4f22462c94824b722b16a9b5 : FAILURE in 6m 10s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/865403b94add49cc888d781d3ed74133 : FAILURE in 6m 09s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/97a1e3bad6a04a56a93f38bdffba2cb1 : FAILURE in 6m 03s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/13e1c947cd704a61a6b7cf29baf29dc0 : FAILURE in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8bd52a5cd483cbd4e6e9e49a12e054d179175f0a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-12-20 16:42:37.000000000","message":"Patch Set 12:\n\nAndy,\nIs it possible that the environment is already running a rabbit broker on 5672?  IOW, there\u0027s two rabbits: one started by the host using default port 5672 and the one our test creates (port 5682)?\n\nIf so the dispatch router would exit immediately, but since the log isn\u0027t available for us to view it\u0027s hard to tell...","accounts_in_message":[],"_revision_number":12},{"id":"4a49f6b3a511e07ee726fe8bf941f9b1ca53450f","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-20 17:55:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fd54b3c095953f8e2b148e54a1d6eb0547e62f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 19:37:09.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/df5df1cca60141abbfbaa9251b25e7ce : SUCCESS in 1h 38m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8471bf9a79c487c9f7ec4763beeace0 : SUCCESS in 1h 40m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/927bede05a304c6fa37e1e52c565a344 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8ef8427f82c4de2a9016fe90765ee4c : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/436f058b9dac4f1c9a804ec42d057aef : FAILURE in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac23389960ab44e1bbee0702139dc675 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db968b66d9b741589e0f66751eecb8f8 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09158ad825b84fe2953a9b78d1bffdba : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae9ac32ea8346a9a959e4c85e154ac9 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02480001af8f4ecdb244c23e5db9c217 : SUCCESS in 4m 00s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/b62fdbf9cff54d0c808a0434da77c70c : SUCCESS in 4m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/5cefbecddb60482e844e36de73759317 : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/447cbac4bb8c42618fa93eaf089a0f2e : SUCCESS in 5m 19s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/feefd72962ac453bb9a4decbea1f4b98 : FAILURE in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"bd898f568706d596858225e2e5ea447da2dabf42","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2019-12-20 20:35:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2d737e0a9726389d2bbecb1a92f2c7ab2cfa42ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 22:08:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7cd37e72285646abbd8da10af6cbd272 : SUCCESS in 1h 32m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1798aa7499224ed9bcae063d89a1a4aa : SUCCESS in 1h 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9ebea0908e4bc282f3b9c903622317 : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b404b95315c48539c0671f2e12e5b76 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a7493d52eb420eb405b426dc9d1ae1 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53e823e58ae7458ba9bf86b1fd2481dd : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6adddb49bc8f4e70b659615d65cd27c7 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98dcd7ef177d47cea330887ae16a7581 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ab04d69f9b4fde8c4e33e20f03600d : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00cb737bf6ac442483c20b80e7ec991a : SUCCESS in 4m 25s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/ae09127e22ad4b58975b77dfd1bc2d46 : SUCCESS in 6m 35s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/be2fa36cce4b4c8b933f2b44ec139b7b : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/78f5964660184e41827851f597a5c7fc : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/4299d03a14af4da8830532df351718da : FAILURE in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a1934f9289696427311e70ad6aa390ef08a1dd11","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-07 14:11:03.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ce4a2350c2cb34412a2a7237f9e40728a0cb0ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:52:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/129e3c1a58844d1580918176129b4990 : SUCCESS in 1h 23m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/225b99990bea42c89e555b42ae6ea093 : SUCCESS in 1h 32m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6058aa6cb91f4eabb489d856ed9c4489 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4165e3fcd0c4e20a1af9c0170ec8d54 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05a50a36f6a43cfa275795921dc21bd : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e005cadce4024c619ae522d6f319544d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a538351a86e049aaaf37f68b97fa5685 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b6377b514684bf2a3964d80f60c488f : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414ae258754a462484dc931f87e89e95 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/835e0aaec3d1404b8da07f1abc58c604 : SUCCESS in 4m 28s\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/8c5fca73f8c04507a03a7b4eef8c70fa : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/99ec8d93c4024ac8b3ae1f368b0c2378 : SUCCESS in 6m 31s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/dc0bfb68e227447ba477f3c4d4571292 : SUCCESS in 5m 35s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/b732d3f3aae048b0bbed14f9c1c7466f : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2473e2e625487f2f406ba594075fb8d01cf67632","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-07 16:50:24.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0196d8bdffff62dd4d78c901da85067d90add1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 16:50:39.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.messaging on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027oslo.messaging-tox-py27-func-scenario01\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-ussuri-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - oslo.messaging-tox-py27-func-scenario01:\n          - oslo.messaging-tox-py27-func-scenario02:\n              voting: false\n          - oslo.messaging-tox-py27-func-scenario03:\n              voting: false\n          - oslo.messaging-tox-py27-func-scenario04:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario01:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario02:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario03:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario04:\n              voting: false\n  \n          - oslo.messaging-src-dsvm-full-rabbit\n          - oslo.messaging-src-dsvm-full-amqp1-hybrid:\n              voting: false\n          - oslo.messaging-src-dsvm-full-amqp1-centos-7:\n              voting: false\n          - oslo.messaging-src-dsvm-full-kafka-hybrid:\n              voting: false\n          - oslo.messaging-src-dsvm-full-kafka-centos-7:\n              voting: false\n  \n          - oslo.messaging-src-grenade-dsvm:\n              voting: false\n          - oslo.messaging-src-grenade-dsvm-multinode:\n              voting: false\n  \n      gate:\n        jobs:\n          - oslo.messaging-tox-py27-func-scenario01\n          - oslo.messaging-src-dsvm-full-rabbit\n\n  in \"openstack/oslo.messaging/.zuul.yaml@master\", line 166, column 3","accounts_in_message":[],"_revision_number":16},{"id":"be0a8b2947f5466ea5675136be6b6d0e9484e74f","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-07 16:54:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5ff77d1d8ae7bab909ebc61b23e40db3ae56d125","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-07 17:39:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"86b77cf622a18a9daa285ed1413502c412721f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:49:35.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8c343ae5e7d644cdaa5a5c50d9b3b557 : SUCCESS in 1h 23m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57ed0c9967ee42898bbd0b6ad4adcb46 : SUCCESS in 1h 18m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c3a8cf9c0449518386bd307e116c93 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f483b66944cf4a5f8611657ea6ff5d18 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd3f2e6f7c14734b7cd1cdf05b51336 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/943bc149576e427a8e80687787f39456 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e97a4f5987d64d358c33eaca2917a583 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e029822790cd436eb5f542a09c082a86 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c7d23d7a7c4eb7873baa1be1cfb3c4 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baf8d3ca6e594581b29d8639b734d491 : SUCCESS in 3m 58s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/94a8677d5a5d445e8bdda3a7bcc5fa52 : SUCCESS in 6m 00s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/1bf86906328c41eaa57ecf55cd2d87ea : SUCCESS in 6m 48s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/615a9602f9934674994e8f58deb54845 : SUCCESS in 8m 47s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/117c9235ef624983bed91718c7412fa3 : SUCCESS in 7m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e9262c0b3df240988d6f841be6b538fc : SUCCESS in 5m 30s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/d98541480b804219bccc1e467214922e : FAILURE in 7m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/514decdfa2544d8cb07b46c6a87674b8 : SUCCESS in 6m 06s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/854d1b1140754e48901cb07211863247 : SUCCESS in 7m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ff3145095c3d46cfba45e01b2dcd47d5 : SUCCESS in 1h 29m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/17c5191257d04c1ba1f3666852f4200e : SUCCESS in 1h 22m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/9b15b3526e394a85abe661781f5904e6 : FAILURE in 13m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f88dca40e22042518c8b51fc69d6fd78 : SUCCESS in 1h 34m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/f0ba6b3c916b4ea1a25105cfbe22ea00 : FAILURE in 11m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8b95dce607044b9cbcc19a3cbd7b446a : SUCCESS in 1h 00m 36s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/fa8df159ffaf4054900b14c1052722e2 : FAILURE in 1h 07m 11s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b16cd3070a9f9fe2618482ba7f2ee4c4f7eff7db","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-12 19:06:21.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"9cc204d3297426dc8a27bfc0ef68ae1b1b42e665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 19:44:02.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/588751213cb541eda753905e24899f94 : FAILURE in 9m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/343f6ef91d044480b37b168ec2d2ab0c : FAILURE in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44dac3378a124c0f807573c1e4d3a2d5 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2dd3f946b54198b2e25f42262a0a3c : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bcefb795a104127871e632fc3a4f51c : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6890c2cf5aea40f1a721f26bd92c21b3 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9636a22e69b4c878c892c92d9eccf42 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35f3f7ad04b542019cde2047e3b73635 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1feb78bdda5443eb9365f73ff301cd9b : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45d2019f002b42f6ab44a95bf3bdfa3b : SUCCESS in 4m 46s\n- oslo.messaging-tox-py27-func-scenario01 https://zuul.opendev.org/t/openstack/build/440318f2ffe74dbf953ad3e4e1bccc9f : SUCCESS in 6m 00s\n- oslo.messaging-tox-py27-func-scenario02 https://zuul.opendev.org/t/openstack/build/af251f1dfd734905bb62e5c05075a85f : FAILURE in 7m 42s (non-voting)\n- oslo.messaging-tox-py27-func-scenario03 https://zuul.opendev.org/t/openstack/build/7d947318ba98470e9f3c7e964f5ca68d : SUCCESS in 6m 28s (non-voting)\n- oslo.messaging-tox-py27-func-scenario04 https://zuul.opendev.org/t/openstack/build/5a844ac50513475d8bacdb9339159f9d : FAILURE in 8m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ad337287cf3b44d0a8f96aad8a8e9bc3 : SUCCESS in 5m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/5cd7c2e1ce614198927760a867b14d72 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/4c125cb101bd4985b9c605e840a23f4e : SUCCESS in 6m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/9dd305bde85b4c8a93b1b724848b9bef : FAILURE in 9m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bd8773da52ab4353aedd6e84840a1c3e : FAILURE in 12m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/acb4402da9d5450aaf1c74b76d1d5160 : FAILURE in 8m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/f20f1cc794e544f29f975f720da0acc7 : FAILURE in 17m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/06184ed905d14b11b114b201a934abe2 : FAILURE in 18m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/da74c01079db45d8bbfa84f0043440d2 : FAILURE in 20m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b0311334a9fa4c90bd71f22ae8a13c87 : FAILURE in 5m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/7ecd45b3211841cb89beaa6481e9cc6f : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d5fbe26a346cc21e1c4b11f8501fe35edb2225ec","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-01-13 14:41:16.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1486c7c77411d4a467e45e97aaf980213ef6432f","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 13:58:02.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6a7ca3784596a218e205c03031004c2013868834","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 13:58:20.000000000","message":"Patch Set 20: Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"c51840b80be39d15a878501305f9f1949e9f070c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 13:58:22.000000000","message":"Patch Set 20: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.messaging on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027oslo.messaging-tox-py36-func-scenario01\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python-jobs\n        - openstack-python3-ussuri-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - oslo.messaging-tox-py36-func-scenario01:\n          - oslo.messaging-tox-py36-func-scenario02:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario03:\n              voting: false\n          - oslo.messaging-tox-py36-func-scenario04:\n              voting: false\n          - oslo.messaging-src-dsvm-full-rabbit\n          - oslo.messaging-src-dsvm-full-amqp1-hybrid:\n              voting: false\n          - oslo.messaging-src-dsvm-full-kafka-hybrid:\n              voting: false\n          - oslo.messaging-src-grenade-dsvm:\n              voting: false\n          - oslo.messaging-src-grenade-dsvm-multinode:\n              voting: false\n  \n      gate:\n        jobs:\n          - oslo.messaging-tox-py36-func-scenario01\n          - oslo.messaging-src-dsvm-full-rabbit\n\n  in \"openstack/oslo.messaging/.zuul.yaml@master\", line 119, column 3","accounts_in_message":[],"_revision_number":20},{"id":"cc870cf680fbe377cfce0c4da1c974de3dc8ec37","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 13:59:37.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"fb95e5003163ff7634d477a284db7412b986792f","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 14:00:06.000000000","message":"Patch Set 21: Workflow-1","accounts_in_message":[],"_revision_number":21},{"id":"a9453c8a6220be831aa7a1b5d884ced0edb36906","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 14:39:20.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"cc4bfb406ea46ad2739615d7f71291c97555c557","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-21 14:40:55.000000000","message":"Patch Set 22: Workflow-1","accounts_in_message":[],"_revision_number":22},{"id":"f3a77a80273a6fb75aa2863cda120e72469492c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 16:35:20.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/388d7f17ee9f4242b09be6ed17d78ee8 : FAILURE in 27m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9f9e8e81256483292baf975a0c1a9ce : SUCCESS in 1h 49m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07cfa956f5af4d0ba8d1026692e6192c : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b3f045cd6ad43098e587cf7152d99c1 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136af1a0f3ea411dbfff45d9ac91c2cf : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10cc9960ced74ebab4a6a0daad04162d : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8fa5c825e34edea0bcd32382839197 : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4acd3baf386941309f032e032c4e6554 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/114134c0a186417fb4c46c7adb450291 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/833c2a5f7a0b4b3084c20f2091cd92fe : SUCCESS in 5m 58s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/79c1a96f31b148eb89b8ee68e8d3ae39 : SUCCESS in 6m 35s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/92196c976df84c36823deacec03cb838 : SUCCESS in 7m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/10239925d9214ccbbce5cded8557de01 : SUCCESS in 7m 12s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/66ab99bd2b63482e94041e1ceeaee0bb : SUCCESS in 7m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2e86f2dea334430b870ea9299deac31b : SUCCESS in 1h 31m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9c81fe397fe34099a97f1e7cdd229919 : FAILURE in 30m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/81e684eb3c0b4d69b974b77c7b82cba5 : FAILURE in 31m 15s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a3fed4fbe6d446709bc04875c99a03c7 : SUCCESS in 1h 18m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5e94bf4d91094137852304d5dc5ba008 : FAILURE in 1h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"4a08a1b344e8a8d6357f60282ce2e036bd348ca5","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-22 14:09:53.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"e05e93133d8afea7455311c33423434eca426f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 16:18:56.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5d0e28825182498aa4282f21e598d5c8 : SUCCESS in 1h 27m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb9dc1c8c221482cb39f063f7b0c7998 : SUCCESS in 1h 25m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4ccc66c089440379e8609bbe9ca1336 : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6af2d955dbd4950b9b72d0f6d31ba97 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699fd46dabe94298930bfc1c1816d968 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16ce6c25baef4c55a8382be611c10ac0 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5db793936bb8425da41232a2d664c62b : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42fd63c70f3d4646aca06ff9633bd119 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e1cc7162454d3ab210b0ef7b95cccd : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2ddc4c4c404dee9be89aaf803e20df : SUCCESS in 6m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/6cbe44716df942c48b4d6b5281b00749 : SUCCESS in 9m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/67b904c4a24f4ca69457a5c085d35012 : FAILURE in 10m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/17db018dd6aa44ec881844875c8eb587 : SUCCESS in 7m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/8a2679f802a942948b2130ba3885d722 : FAILURE in 7m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0e7cb12da73f4e059b45d239588578ed : SUCCESS in 1h 45m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d2425fc699834008a1fdfb27830d5a24 : SUCCESS in 1h 41m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/276102164e0d4e558f28b0b9993cc13f : SUCCESS in 2h 04m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/965a8367d0e54c72b8ff6da35f574f5d : SUCCESS in 1h 10m 59s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/294829499e1143a9bab12da673d6be36 : FAILURE in 1h 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7d7c9233812f903927086aa593fb2cd28e4b89fa","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-01-22 17:47:48.000000000","message":"Patch Set 22: -Workflow","accounts_in_message":[],"_revision_number":22},{"id":"f74aec4df6d19f63ef741bc24d1a3f5ec0096a1c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-23 10:29:44.000000000","message":"Patch Set 22:\n\nHello,\n\nscenario002 and scenario004 failed, the both are kafka based so maybe we have an issue related to the kafka setup...\n\nMaybe I missed something but the given logs are not enough verbose to understand why tests fails, I\u0027ve an idea?\n\nYou definied `set -e` in https://review.opendev.org/#/c/677744/22/tools/setup-scenario-env.sh so maybe something went wrong during `_setup_kafka` and it exit with error and then your script trap the error and stop its execution.\n\nBy example with the wget in `_setup_kafka`, requesting a wrong url or requesting against a server not available could exit in error here.\n\nThoughts?","accounts_in_message":[],"_revision_number":22},{"id":"ceead897cc369f38d58b08e8daa03c51ad8b286c","author":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"date":"2020-02-05 13:00:56.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"c7ba1da6537c816f996b373fd47e5cdfa50bcac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 15:02:32.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2723eb2c4354dcb879fb3c96bbe38ef : SUCCESS in 1h 41m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a37e65e6fa7e4f5c874c72ea1239d25c : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/584b398c0f5a4efc9630b25d7ec39b8d : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ece8e2b4fe495182baf75d2e65b727 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/103989dc1e6e488cb8f66fa9841b6a80 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dfa452c001e4b0eadaebc29d62b5187 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b909116c17f54037a7d8ced2329fba83 : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef37712da771406abba7437ab03d9706 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab06aea257a4598b46a9ba05f164121 : SUCCESS in 5m 23s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/fd6b1af395e14bbebeb944e7ba78cd6b : SUCCESS in 8m 52s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/d24cc101afc147feb90ed04bfb8e825c : FAILURE in 9m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/6b1f047f3d2244728e60a98ab66f83d7 : SUCCESS in 7m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/7de2589b5bff4c48acaf7f3aec5496dd : FAILURE in 9m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c5e6cac3ac214b1d98f1eb58bbbad5a6 : SUCCESS in 1h 37m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e756efbd6e924d998d9d6c20842ed6a6 : SUCCESS in 1h 49m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8acd66396dc044c0abeb2ca25a32e3c6 : SUCCESS in 1h 54m 19s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/efca3af1084f44e5a9153b7aa556bab1 : SUCCESS in 1h 16m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6369c522b1c64c3f8f4920e85b205a46 : FAILURE in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"ea4dafdaeea66ee2501d76b99972ea58abb39655","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-10 14:59:38.000000000","message":"Patch Set 23:\n\n(1 comment)\n\nHello,\n\nSee my questions inline","accounts_in_message":[],"_revision_number":23},{"id":"4b9c4b0ef4a2d788985ab654c01ffa26083d4b96","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-02-13 17:43:42.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)\n\nI\u0027m tentatively +2 this patch as it stands.\n\nThe Kafka tests have been failing prior to this patch, so I wouldn\u0027t necessarily hold this up for them.  I\u0027m good with a follow up patch to address the Kafka issues.\n\nWhat I *do* want is to at least get scenario03 to be voting.  This scenario uses the amqp1 driver which has been inadvertently broken in the past due to the fact that the functional test was non-voting due to instability.\n\nSo I\u0027d like to see this upstream, then fix the kafka stuff afterwards.\n\nThoughts?","accounts_in_message":[],"_revision_number":23},{"id":"6741d982fe6256d57964618b27e41650fcac1e60","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-21 14:12:42.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n\u003e (1 comment)\n \u003e \n \u003e I\u0027m tentatively +2 this patch as it stands.\n \u003e \n \u003e The Kafka tests have been failing prior to this patch, so I\n \u003e wouldn\u0027t necessarily hold this up for them.  I\u0027m good with a follow\n \u003e up patch to address the Kafka issues.\n \u003e \n \u003e What I *do* want is to at least get scenario03 to be voting.  This\n \u003e scenario uses the amqp1 driver which has been inadvertently broken\n \u003e in the past due to the fact that the functional test was non-voting\n \u003e due to instability.\n \u003e \n \u003e So I\u0027d like to see this upstream, then fix the kafka stuff\n \u003e afterwards.\n \u003e \n \u003e Thoughts?\n\nI\u0027m ok with that, let\u0027s go to address them in a follow up patch","accounts_in_message":[],"_revision_number":23},{"id":"1a9d5d0efa4d50bad10c9582d7787aefdde5809f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:13:01.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"f6cf404ec35a6404fc756d7949df74e37b73cd2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 15:52:26.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/721ec4c2a50b46d4a052a8cc550342af : SUCCESS in 1h 26m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/424e44bebc994339b1aedcbcfcb1395f : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2fb3090f444136ae5bdf24b6f53daa : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7059eb958cb64319ae191b79f52856e3 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1ef80d687364584a0cdfa3dc482a617 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbd773b7a6e4729ad309f8168669db3 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55e9f50774348bba9774e909a89442f : SUCCESS in 5m 34s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/e5c1b0e45bc546689141e1fcf57d0ce6 : SUCCESS in 7m 59s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0024f745f63d4c25b89bdefd049df0f9 : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":23},{"id":"55ba31b2caba3baa61e1e3fbb5550e8c44ed735e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 15:52:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"27dfac640f2b8babf8b1dc496886c255dcae516c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 15:58:39.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7ad31b79e54ad5a832b01841dd6f50 : SUCCESS in 3m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbfb81fc3fbb47f194ff7499237a9cda : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"ab78c8e3dde1a598159630059b5c78568a9c7425","revisions":{"e030a397cc5e01a3a0d18758832d4c0c2421eb1e":{"kind":"REWORK","_number":1,"created":"2019-08-21 14:41:03.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/1"}}},"commit":{"parents":[{"commit":"b56380654a4b695a8a27827621b185ee3991ce92","subject":"Introduce RabbitMQ driver documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b56380654a4b695a8a27827621b185ee3991ce92"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e030a397cc5e01a3a0d18758832d4c0c2421eb1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e030a397cc5e01a3a0d18758832d4c0c2421eb1e"}]},"branch":"refs/heads/master"},"1108f34f36424a9658dcff9f5f93b1efdc475811":{"kind":"REWORK","_number":2,"created":"2019-08-21 16:06:47.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/2"}}},"commit":{"parents":[{"commit":"b56380654a4b695a8a27827621b185ee3991ce92","subject":"Introduce RabbitMQ driver documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b56380654a4b695a8a27827621b185ee3991ce92"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 16:06:28.000000000","tz":-240},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1108f34f36424a9658dcff9f5f93b1efdc475811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1108f34f36424a9658dcff9f5f93b1efdc475811"}]},"branch":"refs/heads/master"},"bca1f1db558a3761f16502c8fefde573e07e0ced":{"kind":"REWORK","_number":3,"created":"2019-08-26 14:12:21.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/3"}}},"commit":{"parents":[{"commit":"b56380654a4b695a8a27827621b185ee3991ce92","subject":"Introduce RabbitMQ driver documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b56380654a4b695a8a27827621b185ee3991ce92"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-26 14:12:10.000000000","tz":-240},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bca1f1db558a3761f16502c8fefde573e07e0ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bca1f1db558a3761f16502c8fefde573e07e0ced"}]},"branch":"refs/heads/master"},"bc6f16117689a185f4f67211574baa7e3d1dfd7c":{"kind":"REWORK","_number":4,"created":"2019-08-27 15:35:59.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/4"}}},"commit":{"parents":[{"commit":"b56380654a4b695a8a27827621b185ee3991ce92","subject":"Introduce RabbitMQ driver documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b56380654a4b695a8a27827621b185ee3991ce92"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-27 15:35:48.000000000","tz":-240},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc6f16117689a185f4f67211574baa7e3d1dfd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc6f16117689a185f4f67211574baa7e3d1dfd7c"}]},"branch":"refs/heads/master"},"e5269f101c7afcb3cf52c401269a1ac9ac9df05b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-10-07 12:46:06.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/5"}}},"commit":{"parents":[{"commit":"7734ac1376a1a9285c8245a91cf43599358bfa9d","subject":"Merge \"Align message serialization between drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7734ac1376a1a9285c8245a91cf43599358bfa9d"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-10-07 12:46:05.000000000","tz":-240},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5269f101c7afcb3cf52c401269a1ac9ac9df05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5269f101c7afcb3cf52c401269a1ac9ac9df05b"}]},"branch":"refs/heads/master"},"51b773c677c41771c14165229c1481d8bebbb483":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-04 13:52:56.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/6"}}},"commit":{"parents":[{"commit":"6bca848f5b272149bb32353a62f9d37108fcbe15","subject":"Remove unused variable WAKE_UP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bca848f5b272149bb32353a62f9d37108fcbe15"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-11-04 13:52:55.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/51b773c677c41771c14165229c1481d8bebbb483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/51b773c677c41771c14165229c1481d8bebbb483"}]},"branch":"refs/heads/master"},"2cdf3d74acbe13895700ee94ab907565f92c5481":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-12-16 17:16:22.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/7"}}},"commit":{"parents":[{"commit":"859e0d4eaa45d75eb822f4a6be46c547187c6358","subject":"Revert \"Add RPC incoming and reply log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-16 17:16:21.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2cdf3d74acbe13895700ee94ab907565f92c5481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2cdf3d74acbe13895700ee94ab907565f92c5481"}]},"branch":"refs/heads/master"},"ba1981c187f4c64ad2189232fb256835a9297212":{"kind":"REWORK","_number":8,"created":"2019-12-16 20:44:15.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/8"}}},"commit":{"parents":[{"commit":"859e0d4eaa45d75eb822f4a6be46c547187c6358","subject":"Revert \"Add RPC incoming and reply log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-16 20:43:50.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ba1981c187f4c64ad2189232fb256835a9297212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ba1981c187f4c64ad2189232fb256835a9297212"}]},"branch":"refs/heads/master"},"c20049d62fa730dfbfe8bb87e8ef9ce246faf113":{"kind":"REWORK","_number":9,"created":"2019-12-16 21:07:08.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/9"}}},"commit":{"parents":[{"commit":"859e0d4eaa45d75eb822f4a6be46c547187c6358","subject":"Revert \"Add RPC incoming and reply log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/859e0d4eaa45d75eb822f4a6be46c547187c6358"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-16 21:06:59.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c20049d62fa730dfbfe8bb87e8ef9ce246faf113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c20049d62fa730dfbfe8bb87e8ef9ce246faf113"}]},"branch":"refs/heads/master"},"42cb9fee4ea43d6b6b4bf520bc65e29fd33e14ac":{"kind":"REWORK","_number":10,"created":"2019-12-19 17:42:20.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/10"}}},"commit":{"parents":[{"commit":"4fc0e6241b148c29bc937a6bb2c26fa6775ede89","subject":"Merge \"Remove telemetry checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc0e6241b148c29bc937a6bb2c26fa6775ede89"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-19 17:41:26.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42cb9fee4ea43d6b6b4bf520bc65e29fd33e14ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42cb9fee4ea43d6b6b4bf520bc65e29fd33e14ac"}]},"branch":"refs/heads/master"},"f23f46393e0add22011b5293f805201c907e7f2b":{"kind":"REWORK","_number":11,"created":"2019-12-19 21:32:04.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/11"}}},"commit":{"parents":[{"commit":"4fc0e6241b148c29bc937a6bb2c26fa6775ede89","subject":"Merge \"Remove telemetry checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc0e6241b148c29bc937a6bb2c26fa6775ede89"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-19 21:31:55.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f23f46393e0add22011b5293f805201c907e7f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f23f46393e0add22011b5293f805201c907e7f2b"}]},"branch":"refs/heads/master"},"ca485f41aa7c6de250d4347a587e3b456cd45901":{"kind":"REWORK","_number":12,"created":"2019-12-20 13:31:18.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/12"}}},"commit":{"parents":[{"commit":"4fc0e6241b148c29bc937a6bb2c26fa6775ede89","subject":"Merge \"Remove telemetry checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc0e6241b148c29bc937a6bb2c26fa6775ede89"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-20 13:31:09.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca485f41aa7c6de250d4347a587e3b456cd45901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca485f41aa7c6de250d4347a587e3b456cd45901"}]},"branch":"refs/heads/master"},"e3d81a25474c559a8462dd0ded5b52498075bc79":{"kind":"REWORK","_number":13,"created":"2019-12-20 17:55:12.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/13"}}},"commit":{"parents":[{"commit":"4fc0e6241b148c29bc937a6bb2c26fa6775ede89","subject":"Merge \"Remove telemetry checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc0e6241b148c29bc937a6bb2c26fa6775ede89"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-20 17:55:03.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3d81a25474c559a8462dd0ded5b52498075bc79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3d81a25474c559a8462dd0ded5b52498075bc79"}]},"branch":"refs/heads/master"},"090871a87af1af10896d07d5919b62c4cc3723a6":{"kind":"REWORK","_number":14,"created":"2019-12-20 20:35:28.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/14"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-12-20 20:35:26.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/090871a87af1af10896d07d5919b62c4cc3723a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/090871a87af1af10896d07d5919b62c4cc3723a6"}]},"branch":"refs/heads/master"},"eabef10aaf06c603aff0cbee0e94bed27c534b3c":{"kind":"REWORK","_number":15,"created":"2020-01-07 14:11:03.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/15"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-07 14:11:02.000000000","tz":-300},"subject":"[DNM] Setup backend scenarios for functional tests","message":"[DNM] Setup backend scenarios for functional tests\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eabef10aaf06c603aff0cbee0e94bed27c534b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eabef10aaf06c603aff0cbee0e94bed27c534b3c"}]},"branch":"refs/heads/master"},"ecf4d0ae5c2a0bd6c4db7243a4f1edb3a1cccd57":{"kind":"REWORK","_number":16,"created":"2020-01-07 16:50:24.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/16"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-07 16:50:22.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ecf4d0ae5c2a0bd6c4db7243a4f1edb3a1cccd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ecf4d0ae5c2a0bd6c4db7243a4f1edb3a1cccd57"}]},"branch":"refs/heads/master"},"ed41924f16b581645e6cf926ccb05ea6e240c2d1":{"kind":"REWORK","_number":17,"created":"2020-01-07 16:54:31.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/17"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-07 16:54:23.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ed41924f16b581645e6cf926ccb05ea6e240c2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ed41924f16b581645e6cf926ccb05ea6e240c2d1"}]},"branch":"refs/heads/master"},"dd3143e9a9302367ca8bf86806d4cf37e0b35a65":{"kind":"REWORK","_number":18,"created":"2020-01-07 17:39:25.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/18"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-07 17:39:15.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dd3143e9a9302367ca8bf86806d4cf37e0b35a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dd3143e9a9302367ca8bf86806d4cf37e0b35a65"}]},"branch":"refs/heads/master"},"ab2981e94bd45b99320226de390de5db7f95ae30":{"kind":"REWORK","_number":19,"created":"2020-01-12 19:06:21.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/19"}}},"commit":{"parents":[{"commit":"41a44230434cf9037d57f5e26c1cc73559e198c7","subject":"Merge \"Ignore releasenote cache within git untracked files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41a44230434cf9037d57f5e26c1cc73559e198c7"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-12 19:06:11.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab2981e94bd45b99320226de390de5db7f95ae30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab2981e94bd45b99320226de390de5db7f95ae30"}]},"branch":"refs/heads/master"},"7ace5a680cba9bbb95f356b73129076792c82afa":{"kind":"REWORK","_number":20,"created":"2020-01-21 13:58:02.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/20"}}},"commit":{"parents":[{"commit":"04b2b5d451df7d5aa4047a6b38aa364c400602a3","subject":"Merge \"Adding debug logs on AMQPListener poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/04b2b5d451df7d5aa4047a6b38aa364c400602a3"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-21 13:58:00.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nNote: The removal of python2 jobs are also included patch\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7ace5a680cba9bbb95f356b73129076792c82afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7ace5a680cba9bbb95f356b73129076792c82afa"}]},"branch":"refs/heads/master"},"1f4c6930efcf057ab76f3da8df471b872ad12f06":{"kind":"REWORK","_number":21,"created":"2020-01-21 13:59:37.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/21"}}},"commit":{"parents":[{"commit":"04b2b5d451df7d5aa4047a6b38aa364c400602a3","subject":"Merge \"Adding debug logs on AMQPListener poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/04b2b5d451df7d5aa4047a6b38aa364c400602a3"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-21 13:59:29.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nNote: The removal of python2 jobs are also included patch\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f4c6930efcf057ab76f3da8df471b872ad12f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f4c6930efcf057ab76f3da8df471b872ad12f06"}]},"branch":"refs/heads/master"},"1ba87c97ea57f81b1f58098fb458934a2f42e5d8":{"kind":"REWORK","_number":22,"created":"2020-01-21 14:39:20.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/22"}}},"commit":{"parents":[{"commit":"04b2b5d451df7d5aa4047a6b38aa364c400602a3","subject":"Merge \"Adding debug logs on AMQPListener poll\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/04b2b5d451df7d5aa4047a6b38aa364c400602a3"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-01-21 14:39:09.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nNote: The removal of python2 jobs are also included patch\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ba87c97ea57f81b1f58098fb458934a2f42e5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ba87c97ea57f81b1f58098fb458934a2f42e5d8"}]},"branch":"refs/heads/master"},"ab78c8e3dde1a598159630059b5c78568a9c7425":{"kind":"REWORK","_number":23,"created":"2020-02-05 13:00:56.000000000","uploader":{"_account_id":20523,"name":"Andy Smith","email":"ansmith@redhat.com","username":"ajssmith"},"ref":"refs/changes/44/677744/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/44/677744/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/44/677744/23"}}},"commit":{"parents":[{"commit":"32a1b6e94845b081e9c00d235320b2014cf18d71","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/32a1b6e94845b081e9c00d235320b2014cf18d71"}]}],"author":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2019-08-21 14:39:49.000000000","tz":-240},"committer":{"name":"Andy Smith","email":"ansmith@redhat.com","date":"2020-02-05 13:00:56.000000000","tz":-300},"subject":"Setup backend scenarios for functional tests","message":"Setup backend scenarios for functional tests\n\nThis patchset introduces scenarios for the functional tests to unify the\nsetup and configuration of alternate messaging backends for RPC and\nNotifications. The scenarios are defined to reduce redundancy in\nthe testing of backends and to demonstrate functional correctness\nacross driver combinations.\n\nCurrent driver support:\nrabbit - RPC, Notify\namqp   - RPC\nkafka  - Notify\n\n                  RPC       Notify\n               ---------  ----------\nscenario01       rabbit     rabbit\nscenario02       rabbit     kafka\nscenario03        amqp      rabbit\nscenario04        amqp      kafka\n\nIt is anticipated that additional scenarios will be defined as new\ndrivers are introduced and/or new messaging intermediarites are supported.\n\nNote: The removal of python2 jobs are also included patch\n\nChange-Id: I0f86416623a0b718516147f0660b4df2b74cf867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab78c8e3dde1a598159630059b5c78568a9c7425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ab78c8e3dde1a598159630059b5c78568a9c7425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
